)]}'
{"id":"openstack%2Fskyline-console~868259","triplet_id":"openstack%2Fskyline-console~master~I104495a2b7c9937a9e6e922c4fac00d3f4178235","project":"openstack/skyline-console","branch":"master","topic":"fix-magnum-bugs","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2022-12-21 07:50:09.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2022-12-21 08:57:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I104495a2b7c9937a9e6e922c4fac00d3f4178235","subject":"fix: seprate clusters and templates by different project","status":"MERGED","created":"2022-12-21 05:59:31.000000000","updated":"2022-12-21 08:58:38.000000000","submitted":"2022-12-21 08:57:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868259-fix-magnum-bugs","meta_rev_id":"bef05e01004251a1df1ff7939c7254b7058bc368","_number":868259,"virtual_id_number":868259,"owner":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-21 08:57:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2022-12-21 07:50:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2022-12-21 07:50:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-21 06:45:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-21 07:02:46.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"4534652eb0dfd3771d4c7e0b2ba046a29bc61504","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2022-12-21 05:59:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"afb4cfda05545ecf94c7fbadf7a67a926024e876","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 06:45:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac6a750a621c431788d5894b44a4e5b3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91b8b1fe12b94e4f95b679403eb77895 : SUCCESS in 4m 38s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/2076aaf356154c5ca74af7fa390ad34a : SUCCESS in 6m 07s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/1387aafc63b345c0bdc9276f90df74d7 : SUCCESS in 4m 28s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/2817888fd17c4c56859227b4c9288169 : SUCCESS in 4m 49s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/7cc7f3930eca4892b446c4bd15bf382e : SUCCESS in 44m 32s","accounts_in_message":[],"_revision_number":1},{"id":"10bf1c6ee71c6b8c924d49fc951d1bad0a911110","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-12-21 07:50:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0c719566774222e79e33cc846ee8c90174559ec7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 07:50:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5bfa74c66d7e7cda0aa8a46837fa232e202e76ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 08:57:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca50e041ea1f41eb8c06d7b340a61b84\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/072d4cf183a14e43975ca20e71b98ba1 : SUCCESS in 5m 48s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/300ad395a6254cd096f9421a09f0a6eb : SUCCESS in 10m 05s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/b550d0af2cb6441fb3395ca17b0f214c : SUCCESS in 7m 25s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/bc8589ceb82c4783bdecda2470d65408 : SUCCESS in 8m 39s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/6dcc111827324cdfb91d34f6c9389cec : SUCCESS in 1h 00m 27s","accounts_in_message":[],"_revision_number":1},{"id":"0f6228ae16c31fd675765b66b4ec7caef6faf34c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 08:57:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bef05e01004251a1df1ff7939c7254b7058bc368","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 08:58:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3415db0091c4bc996eeae5cd47c4d3a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf70a096e6da441588cc67f3b99af263 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"583386d2ce194f56890a967640a26db432ea326b","revisions":{"583386d2ce194f56890a967640a26db432ea326b":{"kind":"REWORK","_number":1,"created":"2022-12-21 05:59:31.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/59/868259/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/59/868259/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/59/868259/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/59/868259/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/59/868259/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/59/868259/1"}}},"commit":{"parents":[{"commit":"331c2ddd3be1a69ad95bf35c721019cccd9c61d7","subject":"fix: fix the project column","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/331c2ddd3be1a69ad95bf35c721019cccd9c61d7"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-12-21 05:35:41.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-12-21 05:55:17.000000000","tz":480},"subject":"fix: seprate clusters and templates by different project","message":"fix: seprate clusters and templates by different project\n\n1.use detail api to fetch list in console but no detail in administrator\n2.fix the used quota with clusters detail api\n\nChange-Id: I104495a2b7c9937a9e6e922c4fac00d3f4178235\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/583386d2ce194f56890a967640a26db432ea326b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/583386d2ce194f56890a967640a26db432ea326b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[]}
