)]}'
{"id":"openstack%2Fskyline-console~868951","triplet_id":"openstack%2Fskyline-console~master~Ibc2b6029499dc4b8d1688765d1448c4d9fcad183","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2025-12-21 07:37:08.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"32921":{"account":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"last_update":"2025-12-21 12:26:48.000000000","reason":"Change was submitted"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-12-21 12:26:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibc2b6029499dc4b8d1688765d1448c4d9fcad183","subject":"Fix tox4 error","status":"MERGED","created":"2022-12-31 02:11:07.000000000","updated":"2025-12-21 12:27:39.000000000","submitted":"2025-12-21 12:26:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868951","meta_rev_id":"5f61b6dd538d41793b0abc4ebdfdfa408a5e6d33","_number":868951,"virtual_id_number":868951,"owner":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-21 12:26:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2025-12-21 07:37:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2025-12-14 22:55:18.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2025-12-21 07:37:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"CC":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-31 04:03:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-14 18:18:49.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-12-14 19:17:06.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-12-14 19:17:06.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2025-12-21 09:54:19.000000000","updated_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"CC"}],"messages":[{"id":"5b15c33b91b25cf8a4cc207d89b127da831b1348","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"date":"2022-12-31 02:11:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de077b50a00a839b3e39c7832e28c47cf6b54506","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-31 04:03:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/81caa416cca64f84bfc3be3110f388ed\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d6fedab105b42f8990bf42b73a14c30 : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/545af47afb494672ae8e3e5e8c6491d9 : FAILURE in 3m 11s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/de9988c603a4418084a2882047a93021 : SUCCESS in 9m 46s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/6f8fb04f2d4a4da780f8cbc50f11b287 : SUCCESS in 7m 24s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/77f2c341ce9e49adb4472c20bdb0f515 : SUCCESS in 7m 18s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/6c390b80a2c6453497ee9420be16a8fa : SUCCESS in 1h 41m 54s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/33473591ecbb4745a8b50d23d081489b : SUCCESS in 58m 26s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/c588ca4ea8144654a18d5dbca96de639 : SUCCESS in 53m 59s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/0f155dcc9f754bccbe6cdb71bb9abb85 : SUCCESS in 51m 40s","accounts_in_message":[],"_revision_number":1},{"id":"bb7ba5d403bbd5cbd86ef3794f77b920ba8b617a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"date":"2022-12-31 06:46:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c9d613af07eac0fe22fa6347a73ccf056e20ce0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-31 08:07:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/77b4d81be3614d98afcd911309f5bd25\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9a8b9a97da7457a902ad5537a420efb : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec71a862deb2478d8a20d4873869f640 : SUCCESS in 5m 19s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/3b88918049ad4eb7bea20429e6a67029 : SUCCESS in 10m 36s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/beabff55485d4adc8a031259be813516 : SUCCESS in 7m 24s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/0d5123f725964824aaf6318a69d8e879 : SUCCESS in 6m 49s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/b05d6b06303747e29cef7c604216fd93 : FAILURE in 1h 12m 18s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/fe08d7b219044dc9bba2c04c4f765d15 : SUCCESS in 1h 03m 18s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/8300be4c37264fff98cb5935de3fb1ad : SUCCESS in 58m 33s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/eac3deac1752457e971bc40b24a81552 : SUCCESS in 43m 46s","accounts_in_message":[],"_revision_number":2},{"id":"8f4235abd42bcd815cc0fe3d234558f19c2558ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"date":"2022-12-31 08:21:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"21b6088aef3a042d9c24a3fb01b8777a34a08e6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-31 09:56:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f9a0b08b2c94256b5a632c348878dec\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71a17c7e68614f8ba988571a78ad5095 : SUCCESS in 2m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a6e42355f49412083f678d7880b5dcf : SUCCESS in 4m 45s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/ee432e39a5714485b76e8150f23a8c89 : SUCCESS in 11m 10s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/667f3e138d3743258a2d760aae8f8723 : SUCCESS in 7m 23s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/2cbfa7974cdf4c759b7d9b6de592d4f6 : SUCCESS in 7m 10s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ee90d8deb16d4b0b974bfeb8c1ffa85e : SUCCESS in 1h 27m 33s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/3c972c741cb14bb8b6a4125b97b4f0e2 : SUCCESS in 1h 03m 27s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/47d3cf2315ac4063bf6369f27d803e05 : SUCCESS in 51m 14s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/931b0b5ca1d440538b7d391cf46d30e6 : SUCCESS in 57m 45s","accounts_in_message":[],"_revision_number":3},{"id":"e210bf1598326fdc10cbd99661717af791c22a70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-14 14:59:18.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"ffb947801277f4f1a5c9c8b2f49ad9b1f9887bf6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-14 14:59:22.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"0b689f572bbb4b8c9b317aa2adeff706c06162ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-14 15:52:06.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/29cdb17b3e364626b98585a25ca9d263\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37e9950ab0ef45e5bcddd1f75fe457f8 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a575003ce7784ae28c50bb316250afd1 : SUCCESS in 5m 07s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/6775b5e5e32042bdabb34a47beeb909d : SUCCESS in 4m 53s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/34a33f13eed844869e72f90922facc3f : SUCCESS in 3m 08s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/92d936023de34794bc9a9a4839b1aa1c : SUCCESS in 3m 41s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/7bbc131171ae40f284fc458e8c2a9138 : FAILURE in 51m 36s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/1801020f357143ea9b8783df3ceeeb22 : SUCCESS in 42m 36s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/2607c48e995544eda44e2e3631329ba9 : SUCCESS in 49m 58s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/ad2a3dfc811d411f90a08bbd02a52669 : SUCCESS in 48m 31s","accounts_in_message":[],"_revision_number":5},{"id":"e4ac3a95f09468123057820608b47841f55ed17b","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-14 18:18:49.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f400e170809c53ea5814a7b7d69fc1e18054e80e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-14 19:13:09.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/101106b857e346cc92d8b02f7fa9ffc3\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4e1da688abf4ba389a67a7301b45465 : SUCCESS in 2m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23805c71ffd445c49aae86baa3a71a2d : SUCCESS in 6m 13s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/59fae2afb7344da1ab470be347fa8803 : SUCCESS in 10m 42s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/5dd9f0b658f741f0a8b2b805445db0e3 : SUCCESS in 7m 37s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/922f21c3d33d4fdab34be1c707b8279b : SUCCESS in 8m 26s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/e125b79184cb4f8d9a7ac642be3bcc2a : SUCCESS in 48m 56s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/129087388bec484ab119c28b94a89759 : SUCCESS in 47m 04s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/ec3e9f96354443b598273e76a8cd7d51 : SUCCESS in 49m 08s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/45abf53b47be4786929a6943a1fb5e2a : SUCCESS in 49m 24s","accounts_in_message":[],"_revision_number":5},{"id":"ddc9bd4c974d3b453e5ba4cc679459489025c278","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-14 19:17:06.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"b4d862a2fe422ae4e09cf1fddb36faa31a12fcdb","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-14 22:55:18.000000000","message":"Patch Set 5: -Code-Review","accounts_in_message":[],"_revision_number":5},{"id":"2cc8440db95788e93e29c78fc7350ec8c567c21e","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2025-12-21 07:37:08.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"1b883d6cbf0b473a9653aaed113000c9d10b0839","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-21 07:37:38.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"72048ff142910d03d60092c989255cf5354c9011","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-21 08:47:03.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c92541d38f3b440bbf456f76b9001950\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15cfefdea5b5467ba81204e3179edaae : SUCCESS in 2m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bb119f929ee4b4db598777725a4b105 : SUCCESS in 6m 48s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/e77f99d7ea304d69bfd14a33689d9668 : SUCCESS in 10m 54s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/bf9f45ff54d1478fb42ab9a794111bc0 : SUCCESS in 7m 39s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/052f051792a74db8a5b1b38e68247f39 : SUCCESS in 7m 45s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/e90a02082bb1458ba8cc545674c790a6 : FAILURE in 18m 11s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/9b38aba7a1c247d9b3f274d4643dddc1 : SUCCESS in 1h 04m 27s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/fc93e5742e7d4810b5d783f469521c43 : SUCCESS in 52m 20s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/ab49ec95be8540ddbd0c10c4b72110cb : SUCCESS in 52m 12s","accounts_in_message":[],"_revision_number":5},{"id":"1b42ba2ffed1e20e9995b76ef70e6cfad8908334","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-12-21 09:54:19.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5882a36da18d6f2031649305acbd29b73928d028","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-21 11:23:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cffb27e777142a2971b0d3295b6d79e\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d15d62aaf8ba44eda8d7d3be3280b6b9 : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abcfba3853154151bfc2369142dfeee2 : SUCCESS in 5m 50s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/235caca15f2a4438af5938c6d2aedae4 : SUCCESS in 10m 38s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/60f11976e7d34267b0733fb9db678867 : SUCCESS in 4m 28s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/fd79f679920b42928ddeeb4bd77a0f7e : SUCCESS in 7m 31s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/517befd0992545b1b828bd5e1cf73b30 : SUCCESS in 1h 21m 36s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/5d04b78f5d5b46bdb4bf0b76efc989c6 : SUCCESS in 1h 04m 39s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/0919caa3b347475da3df30c5b5ad4551 : SUCCESS in 31m 09s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/109d5298ef1145e2a4499777c6c3d53d : SUCCESS in 40m 12s","accounts_in_message":[],"_revision_number":5},{"id":"81c67de3479bf073b31801c23125aeb7cabbb24d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-21 11:23:55.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"37466990c300dfec1a4cb4c4b8ba8b967f81edb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-21 12:26:48.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2413f08b8584f16ba44b2a514b0937e\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7af584c7e312411f92df54f63468bec6 : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98ab3018900c4ed7baa5105f82ef6d3d : SUCCESS in 3m 35s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/fffcfbe5be254fbda77c0d883cfc76c5 : SUCCESS in 10m 13s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/b66e1ee20d07483eaa1ab8974d3ae50a : SUCCESS in 8m 19s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/571db60425ac4352ae8d52699a4e9c2e : SUCCESS in 7m 01s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/312de5a5f2b142d39369b19abf9a9243 : SUCCESS in 48m 06s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/e036b34517314337a20201b9d9544b9b : SUCCESS in 58m 29s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/062089aaaf6c41f68c6bc72ab64207e5 : SUCCESS in 49m 27s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/a4526725d73043dba39d9eccbb5fbd6e : SUCCESS in 48m 16s","accounts_in_message":[],"_revision_number":5},{"id":"b498e1d4aa5e016364a6c70d8929ca4e03b8f01a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-21 12:26:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"5f61b6dd538d41793b0abc4ebdfdfa408a5e6d33","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-21 12:27:39.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aea83149ca334e429919a28c945231c9\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c431eb702724e57abac0e2c2e3dee0e : SUCCESS in 37s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/511b70aa722e49629dcf1545f02aab38 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"2a1b61476d33a24e0616dd977e4d0e499120bc99","revisions":{"048027f4b9cb04df35acf09cc5cd91cc8585ce17":{"kind":"REWORK","_number":1,"created":"2022-12-31 02:11:07.000000000","uploader":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"ref":"refs/changes/51/868951/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/51/868951/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/868951/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/868951/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/868951/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/51/868951/1"}}},"commit":{"parents":[{"commit":"583386d2ce194f56890a967640a26db432ea326b","subject":"fix: seprate clusters and templates by different project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/583386d2ce194f56890a967640a26db432ea326b"}]}],"author":{"name":"liyou01","email":"liyou01@inspur.com","date":"2022-12-31 02:09:45.000000000","tz":480},"committer":{"name":"liyou01","email":"liyou01@inspur.com","date":"2022-12-31 02:11:01.000000000","tz":480},"subject":"Fix tox4 error","message":"Fix tox4 error\n\ntox.ini started failing with Tox4 which had some\nincompatible changes. One is passenv where we need\nto pass each value in newline otherwise, it fails with\nerror:\n\n failed with pass_env values cannot contain whitespace,\n use comma to have multiple values in a single line,\n invalid values found.....\n\nOther failure is due to the skipsdist \u003d True.\n\nFixing tox.ini for tox4 changes.\n\nChange-Id: Ibc2b6029499dc4b8d1688765d1448c4d9fcad183\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/048027f4b9cb04df35acf09cc5cd91cc8585ce17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/048027f4b9cb04df35acf09cc5cd91cc8585ce17"}]},"branch":"refs/heads/master"},"0af8dc2307032f313af3fa79c455110b4bbcdba8":{"kind":"REWORK","_number":2,"created":"2022-12-31 06:46:01.000000000","uploader":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"ref":"refs/changes/51/868951/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/51/868951/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/868951/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/868951/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/868951/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/51/868951/2"}}},"commit":{"parents":[{"commit":"583386d2ce194f56890a967640a26db432ea326b","subject":"fix: seprate clusters and templates by different project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/583386d2ce194f56890a967640a26db432ea326b"}]}],"author":{"name":"liyou01","email":"liyou01@inspur.com","date":"2022-12-31 02:09:45.000000000","tz":480},"committer":{"name":"liyou01","email":"liyou01@inspur.com","date":"2022-12-31 06:45:44.000000000","tz":480},"subject":"Fix tox4 error","message":"Fix tox4 error\n\ntox.ini started failing with Tox4 which had some\nincompatible changes. One is passenv where we need\nto pass each value in newline otherwise, it fails with\nerror:\n\n failed with pass_env values cannot contain whitespace,\n use comma to have multiple values in a single line,\n invalid values found.....\n\nOther failure is due to the skipsdist \u003d True.\n\nFixing tox.ini for tox4 changes.\n\nChange-Id: Ibc2b6029499dc4b8d1688765d1448c4d9fcad183\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0af8dc2307032f313af3fa79c455110b4bbcdba8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0af8dc2307032f313af3fa79c455110b4bbcdba8"}]},"branch":"refs/heads/master"},"cfe66b9f72a2a5db153b00fe6ae8d77771c3f08a":{"kind":"REWORK","_number":3,"created":"2022-12-31 08:21:29.000000000","uploader":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"ref":"refs/changes/51/868951/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/51/868951/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/868951/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/868951/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/868951/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/51/868951/3"}}},"commit":{"parents":[{"commit":"583386d2ce194f56890a967640a26db432ea326b","subject":"fix: seprate clusters and templates by different project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/583386d2ce194f56890a967640a26db432ea326b"}]}],"author":{"name":"liyou01","email":"liyou01@inspur.com","date":"2022-12-31 02:09:45.000000000","tz":480},"committer":{"name":"liyou01","email":"liyou01@inspur.com","date":"2022-12-31 08:21:10.000000000","tz":480},"subject":"Fix tox4 error","message":"Fix tox4 error\n\ntox.ini started failing with Tox4 which had some\nincompatible changes. One is passenv where we need\nto pass each value in newline otherwise, it fails with\nerror:\n\n failed with pass_env values cannot contain whitespace,\n use comma to have multiple values in a single line,\n invalid values found.....\n\nOther failure is due to the skipsdist \u003d True.\n\nFixing tox.ini for tox4 changes.\n\nChange-Id: Ibc2b6029499dc4b8d1688765d1448c4d9fcad183\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cfe66b9f72a2a5db153b00fe6ae8d77771c3f08a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cfe66b9f72a2a5db153b00fe6ae8d77771c3f08a"}]},"branch":"refs/heads/master"},"928d03c6d8da24d91846a7c99aa080d50532d6a4":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-12-14 14:59:18.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/51/868951/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/51/868951/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/868951/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/868951/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/868951/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/51/868951/4"}}},"commit":{"parents":[{"commit":"583386d2ce194f56890a967640a26db432ea326b","subject":"fix: seprate clusters and templates by different project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/583386d2ce194f56890a967640a26db432ea326b"}]}],"author":{"name":"liyou01","email":"liyou01@inspur.com","date":"2022-12-31 02:09:45.000000000","tz":480},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-14 14:59:18.000000000","tz":0},"subject":"Fix tox4 error","message":"Fix tox4 error\n\ntox.ini started failing with Tox4 which had some\nincompatible changes. One is passenv where we need\nto pass each value in newline otherwise, it fails with\nerror:\n\n failed with pass_env values cannot contain whitespace,\n use comma to have multiple values in a single line,\n invalid values found.....\n\nOther failure is due to the skipsdist \u003d True.\n\nFixing tox.ini for tox4 changes.\n\nChange-Id: Ibc2b6029499dc4b8d1688765d1448c4d9fcad183\nSigned-off-by: Reason li \u003cliyou01@inspur.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/928d03c6d8da24d91846a7c99aa080d50532d6a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/928d03c6d8da24d91846a7c99aa080d50532d6a4"}]},"branch":"refs/heads/master","description":"Edit commit message"},"2a1b61476d33a24e0616dd977e4d0e499120bc99":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-12-14 14:59:22.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/51/868951/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/51/868951/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/868951/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/868951/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/868951/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/51/868951/5"}}},"commit":{"parents":[{"commit":"6727ac0fe828a4ec763983241a8394b2659778ae","subject":"Merge \"Drop basepython\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6727ac0fe828a4ec763983241a8394b2659778ae"}]}],"author":{"name":"liyou01","email":"liyou01@inspur.com","date":"2022-12-31 02:09:45.000000000","tz":480},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-14 14:59:22.000000000","tz":0},"subject":"Fix tox4 error","message":"Fix tox4 error\n\ntox.ini started failing with Tox4 which had some\nincompatible changes. One is passenv where we need\nto pass each value in newline otherwise, it fails with\nerror:\n\n failed with pass_env values cannot contain whitespace,\n use comma to have multiple values in a single line,\n invalid values found.....\n\nOther failure is due to the skipsdist \u003d True.\n\nFixing tox.ini for tox4 changes.\n\nChange-Id: Ibc2b6029499dc4b8d1688765d1448c4d9fcad183\nSigned-off-by: Reason li \u003cliyou01@inspur.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2a1b61476d33a24e0616dd977e4d0e499120bc99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2a1b61476d33a24e0616dd977e4d0e499120bc99"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
