)]}'
{"id":"openstack%2Fskyline-console~869028","triplet_id":"openstack%2Fskyline-console~master~Ic932f3c9d0bd0d96af2ba01c1f3dfc3bf7b2fb52","project":"openstack/skyline-console","branch":"master","topic":"fix-magnum-bugs","attention_set":{},"removed_from_attention_set":{"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2023-01-04 02:17:33.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-01-04 01:28:37.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2023-01-04 02:17:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic932f3c9d0bd0d96af2ba01c1f3dfc3bf7b2fb52","subject":"fix deleting and detail info showing of cluster","status":"MERGED","created":"2023-01-03 08:36:45.000000000","updated":"2023-01-04 02:19:33.000000000","submitted":"2023-01-04 02:17:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869028-fix-magnum-bugs","meta_rev_id":"f52fc96c00386ec11581faf72370104513d9cfb7","_number":869028,"virtual_id_number":869028,"owner":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-04 02:17:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2023-01-04 01:28:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2023-01-04 01:28:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-03 09:42:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-03 09:43:18.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"4942be9062f0cf3c1cdea5d869d849e410e38cb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-01-03 08:36:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"380cb4abeccbab74ced4e57997b949b1dd5a05b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 09:42:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f2e5d26aaef54fef8dfccda99a1f5b4d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47950f56ec66446fbe4b7a1828ec84cc : FAILURE in 3m 04s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/7aa98b1567cb463883931f70dfdb01ef : SUCCESS in 9m 48s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/40aa5ba75cee4826ae2ca77fc1912894 : SUCCESS in 7m 33s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/2a8008aeac054d79b5db7e6bf5ff35c3 : SUCCESS in 6m 51s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/649052b8169046439a5d8ee6fb67f02e : SUCCESS in 58m 40s","accounts_in_message":[],"_revision_number":1},{"id":"8323e41bf274f6e58c1f31b27888d40de962ce3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-01-03 12:45:10.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d05b421e81d5379876daf72b52812941148df442","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 13:42:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f81dd0765c9c47e4969e3c1731b12460\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/407aa45ac4884087af86880f78d516ad : SUCCESS in 4m 08s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/b34b6bb3ef7f4b58a97d24fb54de5f23 : SUCCESS in 10m 15s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/2c78303f63cf4c3ebe2700868faf6cb5 : SUCCESS in 7m 14s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/70930c5e48df4a4e96ab02a32fd5ba08 : SUCCESS in 6m 42s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/6619145e7ad04ec1a2f71913394d0980 : SUCCESS in 52m 33s","accounts_in_message":[],"_revision_number":2},{"id":"8fa78a739c3c02b899ec8d08587e6c3341455b8f","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-01-04 01:28:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4f5d1a6aa9ea8be7aadd11f440daa51c1aa4a57b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 01:29:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f8d4dd9d2e2b45b09ff7944ba8ffccbead1cccdd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 02:17:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1cd2499f30954e2a908ec3c18c456f21\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93495fd285004de4978133abab2bbebf : SUCCESS in 4m 08s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/025711e4b1ae432bbafd5f38f467df2a : SUCCESS in 6m 45s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/fc8b7ea9dd5b45c1837390479923754d : SUCCESS in 4m 41s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/38fc91a5f479426d9472744b791fd455 : SUCCESS in 4m 41s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/7196daf2ec6c4caa83c334043cacc6a3 : SUCCESS in 47m 36s","accounts_in_message":[],"_revision_number":2},{"id":"9affc382b575cd6f782713667f36ec73bc4d3de8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 02:17:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f52fc96c00386ec11581faf72370104513d9cfb7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 02:19:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4b046c3d8374a2aaecb7c9411f9cbd3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/548668d1f4e2452bbdc4a08f99ac1dca : SUCCESS in 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0568a3ea7c69dad62c401212e31ed5692c5597e6","revisions":{"3ab269d1d301ad3cc9a8fa1ef2a0a040cb4aa21f":{"kind":"REWORK","_number":1,"created":"2023-01-03 08:36:45.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/28/869028/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/28/869028/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/869028/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/869028/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/869028/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/28/869028/1"}}},"commit":{"parents":[{"commit":"583386d2ce194f56890a967640a26db432ea326b","subject":"fix: seprate clusters and templates by different project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/583386d2ce194f56890a967640a26db432ea326b"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-01-03 08:36:14.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-01-03 08:36:14.000000000","tz":480},"subject":"fix deleting and detail info showing of cluster","message":"fix deleting and detail info showing of cluster\n\n1. fix the delete condition of cluster instance\n2. use card to show the health status reason\n3. allowed jump to stack detail from cluster detail page\n4. simplify the template card in cluster detail page\n\nChange-Id: Ic932f3c9d0bd0d96af2ba01c1f3dfc3bf7b2fb52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3ab269d1d301ad3cc9a8fa1ef2a0a040cb4aa21f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3ab269d1d301ad3cc9a8fa1ef2a0a040cb4aa21f"}]},"branch":"refs/heads/master"},"0568a3ea7c69dad62c401212e31ed5692c5597e6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-01-03 12:45:10.000000000","uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/28/869028/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/28/869028/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/869028/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/869028/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/869028/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/28/869028/2"}}},"commit":{"parents":[{"commit":"efc5fd88ee71cabe8e5f0b3b3c37a593848c60c8","subject":"fix: eliminate whitespace in passenv values","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/efc5fd88ee71cabe8e5f0b3b3c37a593848c60c8"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-01-03 08:36:14.000000000","tz":480},"committer":{"name":"Boxiang Zhu","email":"zhu.boxiang@99cloud.net","date":"2023-01-03 12:45:10.000000000","tz":0},"subject":"fix deleting and detail info showing of cluster","message":"fix deleting and detail info showing of cluster\n\n1. fix the delete condition of cluster instance\n2. use card to show the health status reason\n3. allowed jump to stack detail from cluster detail page\n4. simplify the template card in cluster detail page\n\nChange-Id: Ic932f3c9d0bd0d96af2ba01c1f3dfc3bf7b2fb52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0568a3ea7c69dad62c401212e31ed5692c5597e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0568a3ea7c69dad62c401212e31ed5692c5597e6"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[]}
