)]}'
{"id":"openstack%2Fskyline-console~869029","triplet_id":"openstack%2Fskyline-console~master~Ib0af083155a85d079aad254ca2acdc6434675673","project":"openstack/skyline-console","branch":"master","hashtags":[],"change_id":"Ib0af083155a85d079aad254ca2acdc6434675673","subject":"fix: eliminate whitespace in passenv values","status":"MERGED","created":"2023-01-03 09:01:35.000000000","updated":"2023-01-03 11:50:57.000000000","submitted":"2023-01-03 11:49:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869029","meta_rev_id":"b63dec15b5fdf471325b51d637a092e47b003c7b","_number":869029,"virtual_id_number":869029,"owner":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-03 11:49:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-01-03 09:33:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-01-03 09:33:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-03 10:37:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3922dd334fe95e37c19e44a9d37f889a82f8fe51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-01-03 09:01:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f6c6ec2e5211bcb4cb03de197ffbf301af71944","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-01-03 09:33:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6460a4e550b8e701fc183a599cc40a10471c81ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 10:37:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28e91eb4f6fe483196ebb430815b005a\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8b88cbf50114ef798a47e701c741f6c : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a42b76d8fe424130aa67b255b1b3faec : SUCCESS in 4m 35s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/6b27f3d2ba0d428d9bc7d0ec97871169 : SUCCESS in 10m 11s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/23c0e52734a0429ea77826cea43df333 : SUCCESS in 7m 25s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/f8e436ae94224481bfd34a27fac901bf : SUCCESS in 7m 07s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/87a2f774844a40cc9eca2be8dfd32aa8 : SUCCESS in 1h 27m 55s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/dcb7fae1cd094c418adf98192e78b496 : SUCCESS in 1h 05m 21s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/9e16dd1ebffc4b649101d945226a6daf : SUCCESS in 54m 39s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/13e5eef0058e41e2a395c381723228ae : SUCCESS in 1h 12m 13s","accounts_in_message":[],"_revision_number":1},{"id":"31f35f6e281be5cd879cd1293f54776da14c058d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 10:38:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bfc42dccb0e81019a4a663ece02b460ef8db1489","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 11:49:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ef70bf647b347558063117318978db0\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e673e3f3c4e4dcfb41e9ccdbbea3970 : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58614663d4874c6c8f42cda5726d68f7 : SUCCESS in 4m 59s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/c861c8a8c2e04f9896b05b0a653f635c : SUCCESS in 6m 53s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/cb45770a339646fb896b8fdc1c76df16 : SUCCESS in 5m 16s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/e25a679e0ab5420095b5e685eb6df6dd : SUCCESS in 5m 53s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/5bb639b319294a2b94c69a4860a52236 : SUCCESS in 1h 05m 17s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/27e8b6d9c6bc438eb2310fc435834b10 : SUCCESS in 1h 05m 28s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/3a75295ee32948c589dfdedbced1928b : SUCCESS in 56m 03s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/bdddbf872c6a42c09522e58331ff5403 : SUCCESS in 39m 07s","accounts_in_message":[],"_revision_number":1},{"id":"93e207355c4a3fcfb18888d51b370c7da9bc7391","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 11:49:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b63dec15b5fdf471325b51d637a092e47b003c7b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 11:50:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c568b4f4a4246a1af2bdbdd5114c2ca\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59195e6a010642c6af714018ce194295 : SUCCESS in 40s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a6e8ddd99ae418d815d390054f40a11 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"efc5fd88ee71cabe8e5f0b3b3c37a593848c60c8","revisions":{"efc5fd88ee71cabe8e5f0b3b3c37a593848c60c8":{"kind":"REWORK","_number":1,"created":"2023-01-03 09:01:35.000000000","uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/29/869029/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/29/869029/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/29/869029/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/29/869029/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/29/869029/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/29/869029/1"}}},"commit":{"parents":[{"commit":"583386d2ce194f56890a967640a26db432ea326b","subject":"fix: seprate clusters and templates by different project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/583386d2ce194f56890a967640a26db432ea326b"}]}],"author":{"name":"Boxiang Zhu","email":"zhu.boxiang@99cloud.net","date":"2023-01-03 09:01:04.000000000","tz":480},"committer":{"name":"Boxiang Zhu","email":"zhu.boxiang@99cloud.net","date":"2023-01-03 09:01:04.000000000","tz":480},"subject":"fix: eliminate whitespace in passenv values","message":"fix: eliminate whitespace in passenv values\n\neliminate whitespace in passenv values\n\nChange-Id: Ib0af083155a85d079aad254ca2acdc6434675673\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/efc5fd88ee71cabe8e5f0b3b3c37a593848c60c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/efc5fd88ee71cabe8e5f0b3b3c37a593848c60c8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[]}
