)]}'
{"id":"openstack%2Fskyline-console~869162","triplet_id":"openstack%2Fskyline-console~master~Ibb0f49eaa8c8721e54f67ae48e0d7c6ea8ccd260","project":"openstack/skyline-console","branch":"master","topic":"fix-download-data","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-01-05 02:04:19.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2023-01-05 07:38:34.000000000","reason":"\u003cGERRIT_ACCOUNT_33689\u003e replied on the change","reason_account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"}}},"hashtags":[],"change_id":"Ibb0f49eaa8c8721e54f67ae48e0d7c6ea8ccd260","subject":"fix: fix the download data.","status":"MERGED","created":"2023-01-04 09:38:32.000000000","updated":"2023-01-05 10:49:22.000000000","submitted":"2023-01-05 10:48:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869162-fix-download-data","meta_rev_id":"7c60e81b1739361e43a1889f0b35377bb85cdaa4","_number":869162,"virtual_id_number":869162,"owner":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-05 10:48:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2023-01-05 06:04:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2023-01-05 06:04:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-04 10:51:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-04 11:25:57.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"dc03742b876a85192a6b25b1fe8c98b906ec8d87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-01-04 09:38:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f75affb539fdb12d37c9deccc5cdd0375835ab5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 10:51:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d029c4b7a6e6411296a4c7f22ad623e6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fc61b5e0b424f298007ae24b784bbf7 : SUCCESS in 5m 13s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/c77f91b474b848f9abd55e39be5b0b70 : SUCCESS in 8m 24s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/3dc29705d2e046128e95174d053e3d22 : SUCCESS in 4m 53s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/a41f326560f04af093dfec604b0da56f : SUCCESS in 5m 57s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ac78d4f322b44aa0ba2081142f7fdd03 : SUCCESS in 1h 11m 17s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/66f53c8a5fbb4b838b84b64035f85719 : SUCCESS in 46m 52s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/e85da2511c3b4298bd137dfdb63c525f : SUCCESS in 43m 47s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/17e6a503dab146c192e30016b3fcadac : SUCCESS in 42m 35s","accounts_in_message":[],"_revision_number":1},{"id":"50762887e941c3fd8ec3de8b23b5c5ed3d350f67","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-01-05 02:04:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"de420fe08c1106ad945119b9ce0a1012dd160b68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 02:12:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a7672bf0e317184347f1fa8e2864233a059b6a3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 03:43:23.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a124cb26f334eff80b348e8db12affa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f641d897eebe45d5ae561bad520f3cec : SUCCESS in 5m 06s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/54786b798f2544aa9f17aa0329b08a0b : SUCCESS in 9m 03s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/3b0c9204c2aa4c4d825687b7d38d8901 : SUCCESS in 6m 03s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/78896096273a4d7bb1255cff1a5a80e9 : SUCCESS in 6m 57s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/bb7a37660d634595b8800702eed404de : SUCCESS in 1h 25m 25s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/ecf98a0c219948d0bc6d9c664fc1253b : SUCCESS in 1h 00m 18s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/a38fbde72290475eb78a999ee840f961 : FAILURE in 1h 12m 18s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/8a84ac249d874264a01fc882868bad84 : SUCCESS in 59m 46s","accounts_in_message":[],"_revision_number":1},{"id":"e0de061108d5c0e0946c471c1008b9100f7712aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-01-05 03:48:30.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9a860a796e609e187f9ca5ba8aef6da9376bf85d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 04:57:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bacd964975b14cb8a6a74a8b88cace74\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98668a83d28042a7bd105c2944a4b8bd : SUCCESS in 4m 58s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/9daf5185941d44d29260bb5c5a078aa7 : SUCCESS in 7m 14s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/0a1d3b498a08467190ee25a064e9353e : SUCCESS in 5m 57s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/2b81308cd83d4168a41a2ab1d44c8064 : SUCCESS in 6m 01s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/bbf9fad71e944da78c08491d8cc148a9 : SUCCESS in 1h 05m 46s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/583c84f6787545668d8c9184a71f8973 : SUCCESS in 40m 58s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/83f32006ffa04f61805b4ded2d35966f : SUCCESS in 41m 24s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/9b02c31f850446b0a7c07e229a655ace : SUCCESS in 39m 10s","accounts_in_message":[],"_revision_number":2},{"id":"d37b5af8463e1c8bdaf31f68ddbf49db57f66142","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-01-05 06:04:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"88a12472e5010aa0fba9277ee64b2ad0db42c9e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 06:05:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6876e18d9a09bf518f7b90fdd77ba6d41750f289","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 07:31:15.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9a80b0ef7ffd43e1802aa3cbef563774\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b43808ccbb6a4aff9210a42ca6fe426a : SUCCESS in 4m 30s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/a7d2cd4d7de9451cadd5035ed82c5dc8 : SUCCESS in 10m 08s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/0e454c6015124b468fb4497ac3ed57e9 : SUCCESS in 6m 52s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/0fcc625c2bce417d9b3cf5c8331348db : SUCCESS in 6m 40s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/1f36ddf55ec5483598e535beabc305cc : SUCCESS in 1h 21m 37s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/e627c967976b4b6d84486ae4310683a9 : FAILURE in 1h 07m 18s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/5d15d4a2f07749b9abbc4b188060d953 : SUCCESS in 58m 24s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/fd07fd5c502d42b7a1a7ae8da139aa25 : SUCCESS in 50m 14s","accounts_in_message":[],"_revision_number":2},{"id":"db67cd5854e15142cbff23b2a4214c28aa14440c","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-01-05 07:38:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5af41ad43a9f83a364178f3c9cd851d587adbe51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 09:10:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/859d5d4dbbe64364b8ed00c2754900ca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7c39fe4e62c4c46b69830f959b5a016 : SUCCESS in 4m 28s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/15f836c7b2874bcfa19967889466c01c : SUCCESS in 10m 26s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/330ff00d47fc47a2a97dc62d591ceda2 : SUCCESS in 8m 09s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/9a30f410858247e9966cde376653605a : SUCCESS in 7m 32s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/188df7f0fb18438d8629c6b9e537d4c2 : SUCCESS in 1h 26m 36s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/018548e0039341e6b10940fef75b7934 : SUCCESS in 1h 10m 33s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/003d88628fa540649c9f2a10c3f9c16e : SUCCESS in 55m 09s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/123273d122c04255b47fb352e8d9cc8a : SUCCESS in 52m 01s","accounts_in_message":[],"_revision_number":2},{"id":"5f21c1b4554dc53a2be063f65abf11fcb5c435da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 09:11:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"040eb613ec2387a1f45a37099f53b859529eb2f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 10:48:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96a0cfa4f2084d309a5f4e8d8aba6139\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2216799a2ded4630838f1926d39a3426 : SUCCESS in 4m 28s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/c484544b5d014b00865483c5a3f875fd : SUCCESS in 10m 35s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/6ad5d51540024ab5b83c3ded600d8fb1 : SUCCESS in 7m 50s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/a8c34b4f87b240719b9a0a1aa516eeb5 : SUCCESS in 7m 31s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/c863437e9f6a483bb8f8a39ab7b12ef2 : SUCCESS in 1h 30m 41s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/b182e48ca9494a01904162b8e0ffb876 : SUCCESS in 1h 16m 15s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/beddec33269b489babc9715130d7a969 : SUCCESS in 47m 59s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/9b5b2fcb1dc84b18916f58a7e024e9e9 : SUCCESS in 49m 55s","accounts_in_message":[],"_revision_number":2},{"id":"7293ad0312546374f8715a9606038abbb6cc0627","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 10:48:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7c60e81b1739361e43a1889f0b35377bb85cdaa4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 10:49:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ea139fbbcd7425d8ca8edf47530c3a0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b477807904644249daad1af8b5ccb34 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a11a0258c4a0035b6c53c9844e9e4fb343d7cd56","revisions":{"3e609395251592fba319936caef00733b32aa972":{"kind":"REWORK","_number":1,"created":"2023-01-04 09:38:32.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/62/869162/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/62/869162/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/62/869162/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/62/869162/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/62/869162/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/62/869162/1"}}},"commit":{"parents":[{"commit":"0568a3ea7c69dad62c401212e31ed5692c5597e6","subject":"fix deleting and detail info showing of cluster","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0568a3ea7c69dad62c401212e31ed5692c5597e6"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-01-04 09:38:01.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-01-04 09:38:01.000000000","tz":480},"subject":"fix: fix the download data","message":"fix: fix the download data\n\nfix the download data by filter params if use backend request\n\nCloses-Bug: #2001607\nChange-Id: Ibb0f49eaa8c8721e54f67ae48e0d7c6ea8ccd260\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3e609395251592fba319936caef00733b32aa972"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3e609395251592fba319936caef00733b32aa972"}]},"branch":"refs/heads/master"},"a11a0258c4a0035b6c53c9844e9e4fb343d7cd56":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-01-05 03:48:30.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/62/869162/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/62/869162/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/62/869162/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/62/869162/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/62/869162/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/62/869162/2"}}},"commit":{"parents":[{"commit":"0568a3ea7c69dad62c401212e31ed5692c5597e6","subject":"fix deleting and detail info showing of cluster","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0568a3ea7c69dad62c401212e31ed5692c5597e6"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-01-04 09:38:01.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-01-05 03:48:30.000000000","tz":0},"subject":"fix: fix the download data.","message":"fix: fix the download data.\n\nfix the download data by filter params if use backend request\n\nCloses-Bug: #2001607\nChange-Id: Ibb0f49eaa8c8721e54f67ae48e0d7c6ea8ccd260\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a11a0258c4a0035b6c53c9844e9e4fb343d7cd56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a11a0258c4a0035b6c53c9844e9e4fb343d7cd56"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[]}
