)]}'
{"id":"openstack%2Fskyline-console~869443","triplet_id":"openstack%2Fskyline-console~master~I40180a44657ace3419c009f73ab527c99fb483d3","project":"openstack/skyline-console","branch":"master","topic":"fix-zun-ui","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-01-10 03:39:36.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2023-01-10 04:57:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I40180a44657ace3419c009f73ab527c99fb483d3","subject":"feature: support some new params and fix some issues of zun ui","status":"MERGED","created":"2023-01-06 10:14:55.000000000","updated":"2023-01-10 04:58:58.000000000","submitted":"2023-01-10 04:57:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869443-fix-zun-ui","meta_rev_id":"5815099172dc65e555960bb4f490924fc9588f7a","_number":869443,"virtual_id_number":869443,"owner":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-10 04:57:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2023-01-10 03:39:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2023-01-10 03:39:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-06 11:50:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-10 03:22:57.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"08e088e416640400c44c4b82f1884a44bb56ae0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-01-06 10:14:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2fd2ce821042fd2f7416259fcfdb981d2f87cc94","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-01-06 10:16:07.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"152ac1fc892834c25b79b3ccce0ac50a424216c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 11:50:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/052a537937544d6db3ce69dec58748cd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9d1c21caebf4e05af7333b7701eb3fd : SUCCESS in 4m 33s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/b9e51af057a2418981aac2ac468a4aeb : SUCCESS in 10m 26s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/a3a67c8b75f34c0db3732c157ea60a69 : SUCCESS in 7m 30s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/9695fe36a16a453c8120224fc991e680 : SUCCESS in 7m 32s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/db04abd780454257aff03a35d4641e11 : SUCCESS in 1h 27m 33s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/b250c60d9cfb4b2fa535de8f9698a923 : SUCCESS in 44m 28s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/ed52c9d4c1274466ba98fbcacdd201fa : SUCCESS in 40m 52s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/c71aca7ec41e45a29c46d18bf18a9522 : SUCCESS in 38m 11s","accounts_in_message":[],"_revision_number":1},{"id":"3dc6a73be232f06c58e55eba53d06110ec1048dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-01-09 10:46:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1deb8b1c69c81688e30463453ea00737a68cb24a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-01-09 10:58:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b97f625bf48864cff15745411cca93850353c39c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-01-09 11:04:43.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6c713c4c6cc318a710770df9bdf7fef7d01ceaf8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-01-09 11:06:21.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6fa1d87f767826cc1697ea46fa466884af4f4836","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-01-09 11:08:03.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"d7953c3cf769cbe5c20d03dd5adb9c4172f243f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-09 12:40:00.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/951d325d505a4768a7f09e7c62337886\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46fcd2437f96464ba6ba7cd9b927b1c4 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/284c45bc93894107bbc42f981d0a18da : SUCCESS in 5m 00s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/3148937aba6a4233a0de1bdeab2dd422 : FAILURE in 9m 34s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/1f3b3ae7b8934da98786bb998b0416b4 : SUCCESS in 6m 39s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/039f1633c7c7412c904d7e33001417cc : SUCCESS in 6m 51s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/f532f91643564ae08f18f1ee3f9bc489 : SUCCESS in 1h 22m 24s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/86f9d101c724453a912edf492634d897 : FAILURE in 1h 27m 02s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/904418a5f35043948e0f40a8f01e81b5 : SUCCESS in 46m 10s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/650418728e4c46a78dd85fbecd03ca15 : SUCCESS in 51m 17s","accounts_in_message":[],"_revision_number":6},{"id":"53aba1e38233e05d732de8edb55ff1dd38b33f4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-01-10 01:46:02.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"c3e42e144a7febcf5c448ce6686be5621d9e6ad6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-01-10 01:50:32.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"994a40981d15f492489bfc282393cda94eb94941","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 03:38:20.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4db2e7d356cf439486fe5d4c221402f3\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/981277d720f44bd58fd29d2f8a6fb5cc : SUCCESS in 2m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b17769be7c004190be66596f57e1b671 : SUCCESS in 3m 41s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/f25d91d714fa4cf9af417244909c1ac4 : SUCCESS in 9m 45s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/7bfe21ebf0c4435486a5a4a4ef965f52 : SUCCESS in 8m 17s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/2d77d70e9e554d9e88ccd77c3ef63ccb : SUCCESS in 6m 47s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/7474013ddb5446f7a8a328f2c543e2aa : SUCCESS in 1h 41m 23s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/ef0682c9623144b692c78413b5e3e8b9 : SUCCESS in 52m 02s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/5571c54d4c764054a1903b482c39cdaf : SUCCESS in 54m 12s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/88d84fb2adb64c89bbb5aa0a73aa932d : SUCCESS in 45m 29s","accounts_in_message":[],"_revision_number":8},{"id":"2ad7a8eec883b4c5e07a504ed8f89b83ec5eacc5","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-01-10 03:39:36.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"339bd340062c94402a82b6eb06494f4cb38f83dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 03:40:16.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"cfa7bd613d62de9ceb0434329788fee44e718a77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 04:57:56.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ad824d70059431da57865e889ca2c2d\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba42561ae83e4d7698eec5aa9a271df1 : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/031366aeddbd44cead57a3ccac3e31ae : SUCCESS in 5m 04s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/9c01203934b94f2282294294f2df47d4 : SUCCESS in 11m 06s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/357a27ff6ae14ce88ac3eaba2ee82de9 : SUCCESS in 8m 25s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/4b31259686cd45c999b219546a169d90 : SUCCESS in 6m 38s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/04f0fd49a7924b7e9f1c91150dab9f3a : SUCCESS in 1h 10m 22s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/bf7be2be323d4815a7cbaa37f60f82c3 : SUCCESS in 1h 05m 54s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/027330a83b6e4aee9e71fa9de2bf3d15 : SUCCESS in 45m 27s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/3874be5a024840e99ec2fac53288c7fd : SUCCESS in 59m 37s","accounts_in_message":[],"_revision_number":8},{"id":"28c76c5a351e8ee73806d6e9b754bed8d3361efd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 04:57:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"5815099172dc65e555960bb4f490924fc9588f7a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 04:58:58.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7abac91284f64a61a764d1c548331146\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29debfadc2bf41089d80eee90edbc84a : SUCCESS in 39s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96dcd1e3de004f038825d119a7a5aac3 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"5f5b8c3cd53e9d1fe8e8bb0cc1bd2425162ea0e1","revisions":{"2c483d5d32af6bf20ded69c6efd512b7b8c71150":{"kind":"REWORK","_number":1,"created":"2023-01-06 10:14:55.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/43/869443/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/43/869443/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/869443/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/869443/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/869443/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/43/869443/1"}}},"commit":{"parents":[{"commit":"a11a0258c4a0035b6c53c9844e9e4fb343d7cd56","subject":"fix: fix the download data.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a11a0258c4a0035b6c53c9844e9e4fb343d7cd56"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-01-06 10:12:49.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-01-06 10:14:16.000000000","tz":480},"subject":"fix: fix the zun ui","message":"fix: fix the zun ui\n\n1. add health check\n2. add exposed ports\n3. fix the image resource by glance or docker\n4. fix step name from miscellaneous to others\n\nChange-Id: I40180a44657ace3419c009f73ab527c99fb483d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2c483d5d32af6bf20ded69c6efd512b7b8c71150"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2c483d5d32af6bf20ded69c6efd512b7b8c71150"}]},"branch":"refs/heads/master"},"164f9b8198da96ff9448acd9cea70e08f768a57c":{"kind":"REWORK","_number":2,"created":"2023-01-09 10:46:31.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/43/869443/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/43/869443/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/869443/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/869443/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/869443/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/43/869443/2"}}},"commit":{"parents":[{"commit":"e678f9c92bd9f2b39a47e2a8fb5fc86323a9f04d","subject":"Merge \"fix: fix the getValue of yaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e678f9c92bd9f2b39a47e2a8fb5fc86323a9f04d"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-01-06 10:12:49.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-01-09 10:45:52.000000000","tz":480},"subject":"fix: fix the zun ui","message":"fix: fix the zun ui\n\n1. add health check\n2. add exposed ports\n3. add auto remove\n4. add entrypoint\n5. fix the image resource by glance or docker\n6. fix step name from miscellaneous to others\n\nChange-Id: I40180a44657ace3419c009f73ab527c99fb483d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/164f9b8198da96ff9448acd9cea70e08f768a57c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/164f9b8198da96ff9448acd9cea70e08f768a57c"}]},"branch":"refs/heads/master"},"8d742d0375932139bb30b60b4322f4a4c4f15c9c":{"kind":"REWORK","_number":3,"created":"2023-01-09 10:58:23.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/43/869443/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/43/869443/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/869443/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/869443/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/869443/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/43/869443/3"}}},"commit":{"parents":[{"commit":"e678f9c92bd9f2b39a47e2a8fb5fc86323a9f04d","subject":"Merge \"fix: fix the getValue of yaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e678f9c92bd9f2b39a47e2a8fb5fc86323a9f04d"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-01-06 10:12:49.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-01-09 10:56:39.000000000","tz":480},"subject":"fix: fix the zun ui","message":"fix: fix the zun ui\n\n1. add health check\n2. add exposed ports\n3. add auto remove\n4. add entrypoint\n5. add validator to some key-inputs\n6. fix the image resource by glance or docker\n7. fix step name from miscellaneous to others\n\nChange-Id: I40180a44657ace3419c009f73ab527c99fb483d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8d742d0375932139bb30b60b4322f4a4c4f15c9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8d742d0375932139bb30b60b4322f4a4c4f15c9c"}]},"branch":"refs/heads/master"},"5d114b227fd23d2c6ede5af9ed408fb7d4be2f16":{"kind":"REWORK","_number":4,"created":"2023-01-09 11:04:43.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/43/869443/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/43/869443/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/869443/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/869443/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/869443/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/43/869443/4"}}},"commit":{"parents":[{"commit":"e678f9c92bd9f2b39a47e2a8fb5fc86323a9f04d","subject":"Merge \"fix: fix the getValue of yaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e678f9c92bd9f2b39a47e2a8fb5fc86323a9f04d"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-01-06 10:12:49.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-01-09 11:03:43.000000000","tz":480},"subject":"feature: support some new params and fix some issues of zun ui","message":"feature: support some new params and fix some issues of zun ui\n\n1. add health check\n2. add exposed ports\n3. add auto remove\n4. add entrypoint\n5. add validator to some key-inputs\n6. fix the image resource by glance or docker\n7. fix step name from miscellaneous to others\n\nChange-Id: I40180a44657ace3419c009f73ab527c99fb483d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5d114b227fd23d2c6ede5af9ed408fb7d4be2f16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5d114b227fd23d2c6ede5af9ed408fb7d4be2f16"}]},"branch":"refs/heads/master"},"dcb152ede613236e8dd4fcb92604d25d1058b17f":{"kind":"REWORK","_number":5,"created":"2023-01-09 11:06:21.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/43/869443/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/43/869443/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/869443/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/869443/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/869443/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/43/869443/5"}}},"commit":{"parents":[{"commit":"e678f9c92bd9f2b39a47e2a8fb5fc86323a9f04d","subject":"Merge \"fix: fix the getValue of yaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e678f9c92bd9f2b39a47e2a8fb5fc86323a9f04d"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-01-06 10:12:49.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-01-09 11:06:07.000000000","tz":480},"subject":"feature: support some new params and fix some issues of zun ui","message":"feature: support some new params and fix some issues of zun ui\n\n1. add health check\n2. add exposed ports\n3. add auto remove\n4. add entrypoint\n5. add validator to some key-inputs\n6. fix the image resource by glance or docker\n7. fix step name from miscellaneous to others\n\nChange-Id: I40180a44657ace3419c009f73ab527c99fb483d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/dcb152ede613236e8dd4fcb92604d25d1058b17f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/dcb152ede613236e8dd4fcb92604d25d1058b17f"}]},"branch":"refs/heads/master"},"0943b67528409dde404ad9bfcda478b219e65926":{"kind":"REWORK","_number":6,"created":"2023-01-09 11:08:03.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/43/869443/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/43/869443/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/869443/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/869443/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/869443/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/43/869443/6"}}},"commit":{"parents":[{"commit":"e678f9c92bd9f2b39a47e2a8fb5fc86323a9f04d","subject":"Merge \"fix: fix the getValue of yaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e678f9c92bd9f2b39a47e2a8fb5fc86323a9f04d"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-01-06 10:12:49.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-01-09 11:07:29.000000000","tz":480},"subject":"feature: support some new params and fix some issues of zun ui","message":"feature: support some new params and fix some issues of zun ui\n\n1. add health check\n2. add exposed ports\n3. add auto remove\n4. add entrypoint\n5. add validator to some key-inputs\n6. fix the image resource by glance or docker\n7. fix step name from miscellaneous to others\n\nChange-Id: I40180a44657ace3419c009f73ab527c99fb483d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0943b67528409dde404ad9bfcda478b219e65926"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0943b67528409dde404ad9bfcda478b219e65926"}]},"branch":"refs/heads/master"},"b0bc655e461078a5b83c10021d6afb1369d35eca":{"kind":"REWORK","_number":7,"created":"2023-01-10 01:46:02.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/43/869443/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/43/869443/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/869443/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/869443/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/869443/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/43/869443/7"}}},"commit":{"parents":[{"commit":"e678f9c92bd9f2b39a47e2a8fb5fc86323a9f04d","subject":"Merge \"fix: fix the getValue of yaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e678f9c92bd9f2b39a47e2a8fb5fc86323a9f04d"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-01-06 10:12:49.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-01-10 01:22:50.000000000","tz":480},"subject":"feature: support some new params and fix some issues of zun ui","message":"feature: support some new params and fix some issues of zun ui\n\n1. add health check\n2. add exposed ports\n3. add auto remove\n4. add entrypoint\n5. add validator to some key-inputs\n6. fix the image resource by glance or docker\n7. fix step name from miscellaneous to others\n\nChange-Id: I40180a44657ace3419c009f73ab527c99fb483d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b0bc655e461078a5b83c10021d6afb1369d35eca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b0bc655e461078a5b83c10021d6afb1369d35eca"}]},"branch":"refs/heads/master"},"5f5b8c3cd53e9d1fe8e8bb0cc1bd2425162ea0e1":{"kind":"REWORK","_number":8,"created":"2023-01-10 01:50:32.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/43/869443/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/43/869443/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/869443/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/869443/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/869443/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/43/869443/8"}}},"commit":{"parents":[{"commit":"e678f9c92bd9f2b39a47e2a8fb5fc86323a9f04d","subject":"Merge \"fix: fix the getValue of yaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e678f9c92bd9f2b39a47e2a8fb5fc86323a9f04d"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-01-06 10:12:49.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-01-10 01:50:16.000000000","tz":480},"subject":"feature: support some new params and fix some issues of zun ui","message":"feature: support some new params and fix some issues of zun ui\n\n1. add health check\n2. add exposed ports\n3. add auto remove\n4. add entrypoint\n5. add validator to some key-inputs\n6. fix the image resource by glance or docker\n7. fix step name from miscellaneous to others\n\nChange-Id: I40180a44657ace3419c009f73ab527c99fb483d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5f5b8c3cd53e9d1fe8e8bb0cc1bd2425162ea0e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5f5b8c3cd53e9d1fe8e8bb0cc1bd2425162ea0e1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[]}
