)]}'
{"id":"openstack%2Fskyline-console~869793","triplet_id":"openstack%2Fskyline-console~master~Ic53c2eb28ad392736d0f78b23eafd13e4c5d88dd","project":"openstack/skyline-console","branch":"master","topic":"fix-zun-ui","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-01-11 09:31:44.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2023-01-11 10:39:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic53c2eb28ad392736d0f78b23eafd13e4c5d88dd","subject":"feature: support ip address, ports and networks in list and detail","status":"MERGED","created":"2023-01-11 08:28:21.000000000","updated":"2023-01-11 10:40:56.000000000","submitted":"2023-01-11 10:39:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869793-fix-zun-ui","meta_rev_id":"db206cac2bfe4cdc85d548cc6120a7c170200abf","_number":869793,"virtual_id_number":869793,"owner":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-11 10:39:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2023-01-11 09:31:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2023-01-11 09:31:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-11 09:22:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-11 09:29:27.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"98b6a3d34d54e9a5eca1e0fd8753619c10e6fd8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-01-11 08:28:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d87295b2fbd2f2591ae8c3cdc3cb2ce2ee47109","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-01-11 08:32:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5ea155c9e3e6ed1bc4cdf090f88053219e877c0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 09:22:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d57ded48cfad478cb0d2ee9b091952bf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e38e1e12dc1d45ab8af8096515d973e6 : SUCCESS in 5m 05s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/09025abe6f9349528d776aa0552242c8 : SUCCESS in 9m 26s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/1d8773c84e97448d9c795dbb0065a422 : SUCCESS in 6m 19s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/4a22a4569cd74bffb0bf996a8078fdca : SUCCESS in 6m 30s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/9d263c2232804752acee1ff648ecb258 : SUCCESS in 47m 48s","accounts_in_message":[],"_revision_number":2},{"id":"fdc90c1e7e8bc5c6ef375a54f0cab73ef24f548c","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-01-11 09:31:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c5f57af933f3794c10076e04ba76ca00d51cdd64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 09:32:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5daf859af2c988259d8be7f050959e2a01f2095a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 10:39:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00bf958511bd40e7bc64fed7fcb472cb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1df6297f0eea4d84a9eee7bdab54f2d4 : SUCCESS in 4m 23s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/d32e6dbedf3a4358ac0aee347083ef15 : SUCCESS in 9m 42s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/c0efdccb003e40fd967e9fe458f47aa5 : SUCCESS in 8m 14s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/0f10be0336844b1d9c94e024f2e55bc4 : SUCCESS in 6m 51s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/3f48c8b5e2284ea4aea09e0eae95011c : SUCCESS in 1h 00m 22s","accounts_in_message":[],"_revision_number":2},{"id":"97c0f80d7f62241228fb16f94ff33cf3559cb6a2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 10:39:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"db206cac2bfe4cdc85d548cc6120a7c170200abf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 10:40:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd29fea62b4c45fbb5e117fc1967403a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71b39a9c9da345a099d025c7b9335cb1 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a347c373a46204d3d3bedcd11c446a5fab8d9f8d","revisions":{"510f962fcd44660588dd62ce9c3642ddedc3895f":{"kind":"REWORK","_number":1,"created":"2023-01-11 08:28:21.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/93/869793/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/93/869793/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/869793/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/869793/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/869793/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/93/869793/1"}}},"commit":{"parents":[{"commit":"317a81acc44b6e509c5352acb4a4a9b7256d8583","subject":"feature: support search filters and fix image in zun ui","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/317a81acc44b6e509c5352acb4a4a9b7256d8583"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-01-11 08:27:49.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-01-11 08:27:49.000000000","tz":480},"subject":"feature: support ip address, ports and networks into list","message":"feature: support ip address, ports and networks into list\n\n1. show ip address, ports and networks in zun containers list\n2. show ip address and subnets into zun containers detail\n\nChange-Id: Ic53c2eb28ad392736d0f78b23eafd13e4c5d88dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/510f962fcd44660588dd62ce9c3642ddedc3895f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/510f962fcd44660588dd62ce9c3642ddedc3895f"}]},"branch":"refs/heads/master"},"a347c373a46204d3d3bedcd11c446a5fab8d9f8d":{"kind":"REWORK","_number":2,"created":"2023-01-11 08:32:07.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/93/869793/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/93/869793/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/869793/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/869793/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/869793/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/93/869793/2"}}},"commit":{"parents":[{"commit":"317a81acc44b6e509c5352acb4a4a9b7256d8583","subject":"feature: support search filters and fix image in zun ui","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/317a81acc44b6e509c5352acb4a4a9b7256d8583"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-01-11 08:27:49.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-01-11 08:31:40.000000000","tz":480},"subject":"feature: support ip address, ports and networks in list and detail","message":"feature: support ip address, ports and networks in list and detail\n\n1. show ip address, ports and networks in zun containers list\n2. show ip address and subnets into zun containers detail\n\nChange-Id: Ic53c2eb28ad392736d0f78b23eafd13e4c5d88dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a347c373a46204d3d3bedcd11c446a5fab8d9f8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a347c373a46204d3d3bedcd11c446a5fab8d9f8d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[]}
