)]}'
{"id":"openstack%2Fskyline-console~870041","triplet_id":"openstack%2Fskyline-console~master~I2de6f5eab11077e249d96bd5a6cbb88dae6e6101","project":"openstack/skyline-console","branch":"master","topic":"fix-magnum-ui","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-01-13 14:49:37.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2023-01-16 02:48:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2de6f5eab11077e249d96bd5a6cbb88dae6e6101","subject":"fix: fix the magnum template detail info","status":"MERGED","created":"2023-01-13 05:04:07.000000000","updated":"2023-01-16 02:50:31.000000000","submitted":"2023-01-16 02:48:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"870041-fix-magnum-ui","meta_rev_id":"2e313262c412a26e564fef6d860a65c15a52cf75","_number":870041,"virtual_id_number":870041,"owner":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-16 02:48:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2023-01-16 01:57:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2023-01-16 01:57:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-13 06:10:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-13 06:11:05.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"44b2188123a561df3d13834a5f69662c181348b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-01-13 05:04:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"462adcba76dc085e3f0bb8b4da23ddddbad4df5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 06:10:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c05d93bda0d4da6bf955851845bc785\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2420bc2bf80d4e94946f9deec1ddbdc9 : SUCCESS in 4m 13s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/aead56e518034f5e91cd8dcd471efe62 : SUCCESS in 10m 54s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/f5629597eb7445f78ee193ad6aad11fc : SUCCESS in 7m 53s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/b96f88b9c67048a88885b0a22950e76b : SUCCESS in 7m 56s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/d89de801908249959e2fb23c750ddbd9 : SUCCESS in 1h 00m 30s","accounts_in_message":[],"_revision_number":1},{"id":"95272389e1f5b467633c4df8148070c5bdabfe24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-01-13 10:12:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"48250abfde3227814d38274ea938c44f6bcd09eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-01-13 10:22:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b817769c07defdbeb3da0214203a1e1b35011239","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-01-13 10:28:46.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"78100a5afded613e95d0b49b214546125c5f5b5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 11:39:09.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c53d05a76442405698ed36006bbc5c08\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4ab8743121e498987c9864e5de6da9f : SUCCESS in 5m 59s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/4b5fc0f2ec4d4dfcafc1a98418d5682b : SUCCESS in 7m 40s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/a86f21910b6c46d89fce85fbd9b30b6f : SUCCESS in 5m 37s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/7e3ba1558766499ab8b5f6f08a1b6f6f : SUCCESS in 5m 38s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/30cc4d1130254d509c0717b508169816 : FAILURE in 1h 08m 22s","accounts_in_message":[],"_revision_number":4},{"id":"aae4723cd606226067453c6c4f74ca12e6dc0be1","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-01-13 13:25:03.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9bd465da687841dd733d118fbf44b50fe978ef24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 14:16:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c29ad968ed5f46dbb0158b412e829798\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c840469cad7472b9f1f5b430ee51755 : SUCCESS in 4m 11s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/1da3f40120384e01a44e246eb59c9e84 : SUCCESS in 7m 47s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/00ce5582b63849349eb82372e1e40e53 : SUCCESS in 7m 02s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/4c9cf26fba074e329fe963a505bc29d0 : SUCCESS in 6m 41s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/e2a210da2eff42b19faf007fb80d3119 : SUCCESS in 45m 57s","accounts_in_message":[],"_revision_number":4},{"id":"c3ec55aee5b5a25434a6283e0b2799cb084639ee","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-01-13 14:49:37.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"bfa3f9768860d3ce98fe5787f405c9baa26df1c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 14:50:12.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"032990444b035795c416fcaf481dcf39a1c84283","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-01-13 15:17:51.000000000","message":"Patch Set 4: Code-Review-1 -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"718636a7ba991b6e9ad48ce20dc37c0a5450374e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 15:18:34.000000000","message":"Patch Set 4: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":4},{"id":"cb4ff698b638a0baa1dbbd20fc846472065db32f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-01-13 16:22:28.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6bb35f7796705c0a1172af9b501c1fdc15f667d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 17:17:52.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93f4ddcb8b554f00ba61abaac5eddbaa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/294b56d7e7f34c82bfe5bdca66afe247 : SUCCESS in 4m 58s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/2fe9c384b07d48648a62d1972f07ef71 : SUCCESS in 13m 43s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/70343cc636fd44c381fc7cd8e8e5a0ee : SUCCESS in 6m 43s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/1dfbf5918d7747dc81d4367dc4a738ae : SUCCESS in 9m 23s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/db13b02e94194937ac82b685cbaade73 : SUCCESS in 53m 09s","accounts_in_message":[],"_revision_number":5},{"id":"33beebfd317655baf6227e309e07949a325e30eb","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-01-16 01:57:47.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"66f56f7338429ca5702a7f0bdce5bf89308f4334","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 01:58:20.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"856308ec476ad3422050296a8733a604a0b363ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 02:48:48.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e987ed962f75481a9eeba92c90378a02\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/302fed02722c40d7bd5b5ca7d1ca8671 : SUCCESS in 4m 27s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/abafdceea40740248e0bb1d0ef6cf78f : SUCCESS in 8m 03s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/7fe9b0425fd143e3a19a7859747efa6f : SUCCESS in 6m 11s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/3a9a0c56975f4ce4a772a216e14e15dd : SUCCESS in 5m 44s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/9c3d3463d30946ecba33f735fd12befc : SUCCESS in 44m 28s","accounts_in_message":[],"_revision_number":5},{"id":"6363ba9fdf2861342548a33f4dc4960bc633ffbb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 02:48:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"2e313262c412a26e564fef6d860a65c15a52cf75","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 02:50:31.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc27c487dae6443b8c4dd9a952d14edd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d02ecef3656456793b95750953c8ebd : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"6c34ed67f03cd1f72225f82fa1481252cecf96b0","revisions":{"bcd7b42cb91381d12b02bf8bfedfe5ad369e9c56":{"kind":"REWORK","_number":1,"created":"2023-01-13 05:04:07.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/41/870041/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/41/870041/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/41/870041/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/41/870041/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/41/870041/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/41/870041/1"}}},"commit":{"parents":[{"commit":"a347c373a46204d3d3bedcd11c446a5fab8d9f8d","subject":"feature: support ip address, ports and networks in list and detail","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a347c373a46204d3d3bedcd11c446a5fab8d9f8d"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-01-13 05:03:44.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-01-13 05:03:44.000000000","tz":480},"subject":"fix: fix the magnum template detail info","message":"fix: fix the magnum template detail info\n\nIf the template\u0027s flavor, network and other attributes are deleted, set it to null\n\nChange-Id: I2de6f5eab11077e249d96bd5a6cbb88dae6e6101\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/bcd7b42cb91381d12b02bf8bfedfe5ad369e9c56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/bcd7b42cb91381d12b02bf8bfedfe5ad369e9c56"}]},"branch":"refs/heads/master"},"2ce9b7ac7c99468397f900253d147baa8887a444":{"kind":"REWORK","_number":2,"created":"2023-01-13 10:12:02.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/41/870041/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/41/870041/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/41/870041/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/41/870041/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/41/870041/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/41/870041/2"}}},"commit":{"parents":[{"commit":"a347c373a46204d3d3bedcd11c446a5fab8d9f8d","subject":"feature: support ip address, ports and networks in list and detail","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a347c373a46204d3d3bedcd11c446a5fab8d9f8d"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-01-13 05:03:44.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-01-13 10:09:24.000000000","tz":480},"subject":"fix: fix the magnum template detail info","message":"fix: fix the magnum template detail info\n\n1. If the template\u0027s flavor, network and other attributes are deleted, set it to null\n2. Fix default network if network are delete when create cluster\n\nChange-Id: I2de6f5eab11077e249d96bd5a6cbb88dae6e6101\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2ce9b7ac7c99468397f900253d147baa8887a444"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2ce9b7ac7c99468397f900253d147baa8887a444"}]},"branch":"refs/heads/master"},"678b31eef41a71ef462771f42a99acfa9714c1d3":{"kind":"REWORK","_number":3,"created":"2023-01-13 10:22:23.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/41/870041/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/41/870041/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/41/870041/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/41/870041/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/41/870041/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/41/870041/3"}}},"commit":{"parents":[{"commit":"a347c373a46204d3d3bedcd11c446a5fab8d9f8d","subject":"feature: support ip address, ports and networks in list and detail","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a347c373a46204d3d3bedcd11c446a5fab8d9f8d"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-01-13 05:03:44.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-01-13 10:22:06.000000000","tz":480},"subject":"fix: fix the magnum template detail info","message":"fix: fix the magnum template detail info\n\n1. If the template\u0027s flavor, network and other attributes are deleted, set it to null\n2. Fix default network if network are delete when create cluster\n\nChange-Id: I2de6f5eab11077e249d96bd5a6cbb88dae6e6101\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/678b31eef41a71ef462771f42a99acfa9714c1d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/678b31eef41a71ef462771f42a99acfa9714c1d3"}]},"branch":"refs/heads/master"},"cc981826361634189ece6d01bf039d778e8afed9":{"kind":"REWORK","_number":4,"created":"2023-01-13 10:28:46.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/41/870041/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/41/870041/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/41/870041/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/41/870041/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/41/870041/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/41/870041/4"}}},"commit":{"parents":[{"commit":"a347c373a46204d3d3bedcd11c446a5fab8d9f8d","subject":"feature: support ip address, ports and networks in list and detail","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a347c373a46204d3d3bedcd11c446a5fab8d9f8d"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-01-13 05:03:44.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-01-13 10:28:33.000000000","tz":480},"subject":"fix: fix the magnum template detail info","message":"fix: fix the magnum template detail info\n\n1. If the template\u0027s flavor, network and other attributes are deleted, set it to null\n2. Fix default network if network are delete when create cluster\n\nChange-Id: I2de6f5eab11077e249d96bd5a6cbb88dae6e6101\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cc981826361634189ece6d01bf039d778e8afed9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cc981826361634189ece6d01bf039d778e8afed9"}]},"branch":"refs/heads/master"},"6c34ed67f03cd1f72225f82fa1481252cecf96b0":{"kind":"REWORK","_number":5,"created":"2023-01-13 16:22:28.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/41/870041/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/41/870041/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/41/870041/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/41/870041/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/41/870041/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/41/870041/5"}}},"commit":{"parents":[{"commit":"a347c373a46204d3d3bedcd11c446a5fab8d9f8d","subject":"feature: support ip address, ports and networks in list and detail","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a347c373a46204d3d3bedcd11c446a5fab8d9f8d"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-01-13 05:03:44.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-01-13 16:22:07.000000000","tz":480},"subject":"fix: fix the magnum template detail info","message":"fix: fix the magnum template detail info\n\n1. If the template\u0027s flavor, network and other attributes are deleted, set it to null\n2. Fix default network if network are delete when create cluster\n\nChange-Id: I2de6f5eab11077e249d96bd5a6cbb88dae6e6101\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6c34ed67f03cd1f72225f82fa1481252cecf96b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6c34ed67f03cd1f72225f82fa1481252cecf96b0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[]}
