)]}'
{"id":"openstack%2Fskyline-console~872073","triplet_id":"openstack%2Fskyline-console~master~Ibd81bc6331fa09a3c70fcfc6f1f72ccc3e6f9c37","project":"openstack/skyline-console","branch":"master","topic":"modal-action","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-01-29 11:58:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibd81bc6331fa09a3c70fcfc6f1f72ccc3e6f9c37","subject":"feat: add attributes to the ModalAction","status":"MERGED","created":"2023-01-29 08:46:35.000000000","updated":"2023-01-29 11:59:10.000000000","submitted":"2023-01-29 11:58:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872073-modal-action","meta_rev_id":"7764eccba0eafab1eea3c0a8648d5e6a75d8ed2e","_number":872073,"virtual_id_number":872073,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-29 11:58:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-01-29 10:20:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-01-29 10:20:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-29 10:04:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-29 10:20:44.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"15e67363f5b64c2fadddc80e2bc1ce88a6a30699","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-01-29 08:46:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08b1a7b51fb940d13aeb196966a8e5bb79f624c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-29 10:04:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dee962b917aa45f1804667a060a5760c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c1ad6be534a430f975049691225220a : SUCCESS in 5m 53s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/ca2154c66c3240b5a97337f6c9071b9c : SUCCESS in 8m 08s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/63df10a3319744c48da41187892df167 : SUCCESS in 6m 46s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/7b3dae56f4f944d78d7dba76a2a61144 : SUCCESS in 7m 18s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ea8641ad103a4527ae4116bc6dc81282 : SUCCESS in 1h 12m 49s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/99125c18a07246928adb1f8274f1e3f3 : SUCCESS in 53m 32s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/51642b096be74f6588dc26547f0810bc : SUCCESS in 44m 25s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/0407e42a776c4600a361d866fd0bc399 : SUCCESS in 45m 42s","accounts_in_message":[],"_revision_number":1},{"id":"a28b6446cf35b9597b8abd3a78849c27134dba6a","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-01-29 10:20:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"53480986cd13e8bae96efde1bcf5d7feb8b24aba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-29 10:21:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"45bfaaee7a1b406f5b2c1dbccb36c0ce28dbd4ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-29 11:58:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f16715811a145eea3fa1e76a4a485e0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/033a7ce1466f45e7a290af5d4ae1a314 : SUCCESS in 5m 07s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/20a3f21c9e264daab70d87d138f6fcd8 : SUCCESS in 11m 18s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/c8b867637818494b82941d85e81e3415 : SUCCESS in 7m 37s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/83de6bf515074d8cb0655092bf421a6e : SUCCESS in 7m 51s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/6793d8fd4b624adca6cd2a202fddeda3 : SUCCESS in 1h 30m 26s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/6624523c4e4f44a794d2b3a113dddcda : SUCCESS in 1h 07m 01s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/3c76a3b29d344eef99e9119473e5b4d3 : SUCCESS in 58m 59s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/4f07e8cc982545bbb7554db6e94c2f0f : SUCCESS in 59m 00s","accounts_in_message":[],"_revision_number":1},{"id":"1ae08969920ac347f0769602ed15504944d0ab62","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-29 11:58:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7764eccba0eafab1eea3c0a8648d5e6a75d8ed2e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-29 11:59:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/010aaabf82e6450394345120a10c48c2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f875a0bb08744c193e063ed2e4e96ee : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2ebb26942a6e1e92de8d5d890cc4f3dfe0b40df4","revisions":{"2ebb26942a6e1e92de8d5d890cc4f3dfe0b40df4":{"kind":"REWORK","_number":1,"created":"2023-01-29 08:46:35.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/73/872073/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/73/872073/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/73/872073/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/73/872073/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/73/872073/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/73/872073/1"}}},"commit":{"parents":[{"commit":"d80c23b202a30fc2551d00aaa0dbea2ca9c0ab69","subject":"fix: fix the title of zun container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d80c23b202a30fc2551d00aaa0dbea2ca9c0ab69"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-01-29 08:46:15.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-01-29 08:46:15.000000000","tz":480},"subject":"feat: add attributes to the ModalAction","message":"feat: add attributes to the ModalAction\n\nAdd the onFinishAction and onCancelAction which are from the ListPage component to the ModalAction, add the modalProps to the ModalAction, these attritubes will make custom operations in the ModalAction to easily control the ListPage\n\nChange-Id: Ibd81bc6331fa09a3c70fcfc6f1f72ccc3e6f9c37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2ebb26942a6e1e92de8d5d890cc4f3dfe0b40df4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2ebb26942a6e1e92de8d5d890cc4f3dfe0b40df4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[]}
