)]}'
{"id":"openstack%2Fskyline-console~872380","triplet_id":"openstack%2Fskyline-console~master~I6ae88611f4431da4e63ed1c65c37b5fd04568373","project":"openstack/skyline-console","branch":"master","topic":"bug-2004462","attention_set":{},"removed_from_attention_set":{"34589":{"account":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"last_update":"2023-02-01 10:55:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6ae88611f4431da4e63ed1c65c37b5fd04568373","subject":"fix: fix the error of compute service state","status":"MERGED","created":"2023-02-01 06:48:14.000000000","updated":"2023-02-01 10:56:34.000000000","submitted":"2023-02-01 10:55:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872380-bug-2004462","meta_rev_id":"97462544970d31e0383580caf201ee74c070ca67","_number":872380,"virtual_id_number":872380,"owner":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-01 10:55:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2023-02-01 10:01:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2023-02-01 10:01:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-01 07:56:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-01 10:01:45.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"eaf5a6f69d949fbaa768ce8e9ae1c4b84e2992d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2023-02-01 06:48:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f2a9f3d2cff4a7f32d09cf0531ba753d436eea67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 07:56:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f495b17abd804e948b415712daa4a474\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bf8f0908f4f4b529a8107f923faa2a6 : SUCCESS in 4m 28s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/d6a9b87679ab439e8f4f10f8f29ec7b7 : SUCCESS in 9m 41s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/e8a5a129727444ecaca8ef8c87291027 : SUCCESS in 7m 28s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/19e47580973e4cfc87808870773be4a2 : SUCCESS in 7m 31s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/cafb8450e18940fbb3100ae4b9585ee4 : SUCCESS in 1h 02m 31s","accounts_in_message":[],"_revision_number":1},{"id":"ae41544eb1a0f07edac6948c709cc352eeda0f28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2023-02-01 08:33:24.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"81e6a6e5cac92eb4fd919f0430729e3e5fe57bae","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2023-02-01 08:34:42.000000000","message":"Topic set to bug-2004462","accounts_in_message":[],"_revision_number":2},{"id":"b411b10527b253f85b8a7c22cf169156a38e1185","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 09:45:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec9feb95a3fa48ef938adddb06c64316\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec85a119f3794d3ca3e4c3755948135b : SUCCESS in 5m 21s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/315faa1de17b40f8bc9219aea3d141c1 : SUCCESS in 9m 38s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/82071ea1feb34939bf2448b3412c904c : SUCCESS in 7m 46s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/249e69bdd4b24825a17bccf76f962095 : SUCCESS in 7m 26s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/49ec43f61f5a49eab7945713a44b9890 : SUCCESS in 1h 02m 06s","accounts_in_message":[],"_revision_number":2},{"id":"3b179b202d6e2c0b99453f2c5896fae8b1a97a12","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-02-01 10:01:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0b4e93997e303a3c2af93d147f70d0d7010c308f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 10:02:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a9f7facdb6d169a37dd759e642db8b6a97f42c9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 10:55:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bc50150b96548489d3a34fad395988c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a174acec40f44b6aada36893b7e914a : SUCCESS in 4m 57s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/02353f3f6b664a119e4fdb46c519da74 : SUCCESS in 10m 22s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/db591c563dc64123a5e517f76d818811 : SUCCESS in 7m 44s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/77d9b33f27d94042bbf972f4bf816277 : SUCCESS in 7m 03s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/3bca03c84bab42249734dcd70b8413cc : SUCCESS in 52m 45s","accounts_in_message":[],"_revision_number":2},{"id":"bc5cd630ba677ed472b32a779cde147b254ca067","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 10:55:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"97462544970d31e0383580caf201ee74c070ca67","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 10:56:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fa88293c03a408caf757388344e2607\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52bd055c6e514836bfb045d7f492dd3f : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"021bde48ef91d76c4da0806ca572645f006dd0b8","revisions":{"8e9cfbdc954313e88dc37aaf9795ef6ff5621f25":{"kind":"REWORK","_number":1,"created":"2023-02-01 06:48:14.000000000","uploader":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"ref":"refs/changes/80/872380/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/80/872380/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/872380/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/872380/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/872380/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/80/872380/1"}}},"commit":{"parents":[{"commit":"fbc8fde8089490f9054c02e933e07444dd88d4a1","subject":"Merge \"feat: add attributes to the ModalAction\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/fbc8fde8089490f9054c02e933e07444dd88d4a1"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2023-02-01 06:41:38.000000000","tz":480},"committer":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2023-02-01 06:41:38.000000000","tz":480},"subject":"fix: fix the error of compute service state","message":"fix: fix the error of compute service state\n\nFix the error of compute service state on Platform Overview page\n\nChange-Id: I6ae88611f4431da4e63ed1c65c37b5fd04568373\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8e9cfbdc954313e88dc37aaf9795ef6ff5621f25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8e9cfbdc954313e88dc37aaf9795ef6ff5621f25"}]},"branch":"refs/heads/master"},"021bde48ef91d76c4da0806ca572645f006dd0b8":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-02-01 08:33:24.000000000","uploader":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"ref":"refs/changes/80/872380/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/80/872380/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/872380/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/872380/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/872380/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/80/872380/2"}}},"commit":{"parents":[{"commit":"fbc8fde8089490f9054c02e933e07444dd88d4a1","subject":"Merge \"feat: add attributes to the ModalAction\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/fbc8fde8089490f9054c02e933e07444dd88d4a1"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2023-02-01 06:41:38.000000000","tz":480},"committer":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2023-02-01 08:32:36.000000000","tz":480},"subject":"fix: fix the error of compute service state","message":"fix: fix the error of compute service state\n\nFix the error of compute service state on Platform Overview page\n\nCloses-Bug: #2004462\nChange-Id: I6ae88611f4431da4e63ed1c65c37b5fd04568373\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/021bde48ef91d76c4da0806ca572645f006dd0b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/021bde48ef91d76c4da0806ca572645f006dd0b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[]}
