)]}'
{"id":"openstack%2Fskyline-console~872625","triplet_id":"openstack%2Fskyline-console~master~I4740c377107a4e8687667c3e24578e1f7846aaba","project":"openstack/skyline-console","branch":"master","topic":"fix-zun-ui","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-02-03 08:25:16.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2023-02-03 09:08:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4740c377107a4e8687667c3e24578e1f7846aaba","subject":"fix: Modify label names to make them consistent","status":"MERGED","created":"2023-02-03 06:53:44.000000000","updated":"2023-02-03 09:09:49.000000000","submitted":"2023-02-03 09:08:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872625-fix-zun-ui","meta_rev_id":"ea787f39a417ad2850ee6895efa776e16858876d","_number":872625,"virtual_id_number":872625,"owner":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-03 09:08:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2023-02-03 08:25:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2023-02-03 08:25:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-03 07:44:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-03 08:18:23.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"a2648a3c5ef8b354e8bac92b4488070bbc3aab6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-02-03 06:53:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4dbd9d900f7b2da3f8bb4451da69411e089d320d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 07:44:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86d9751e219e46899f92251855243382\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/839e12c20fb14edd8dc4a026ab3cf3d6 : SUCCESS in 6m 58s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/4e0248d83d6b4a02a82ed7adf4100e6c : SUCCESS in 11m 52s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/cacca5ce8daf4e918d18c4742bb1268f : SUCCESS in 6m 18s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/162bbef3dc0e46079aa40863d32d0109 : SUCCESS in 5m 00s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/a3cfe3331ec7465797cf5eb2846858e1 : SUCCESS in 48m 55s","accounts_in_message":[],"_revision_number":1},{"id":"6364e11b152f9589545963b827117817e27253e8","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-02-03 08:25:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5fa1513d976d00752e2708159e60dcea6f19b57e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 08:25:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e845768e6fe92170c9a3c655dea4f1f544d67f68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 09:08:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be1564a9f0e24704b3fb90254fde4acb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53d92e18bd624b22ba386de708f948db : SUCCESS in 4m 42s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/982d67028996452dbac4a30d1d152a8b : SUCCESS in 10m 00s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/241bf6c091a04a39b07ca4d34b5c1758 : SUCCESS in 5m 23s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/11a770a4c98e48c988345454993e2021 : SUCCESS in 6m 25s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/b15bf16535d149e8bb1f6417fb674f1d : SUCCESS in 41m 27s","accounts_in_message":[],"_revision_number":1},{"id":"a9b2780226c9c51ebc018e76cecfd0f4d26b0b6f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 09:08:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ea787f39a417ad2850ee6895efa776e16858876d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 09:09:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5867b8b8f065441f95e43657334c45cd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b317d5671a0d4030be3c2dc4538150fe : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5a45f54df369d37b9d80a1f86086e9406876072b","revisions":{"5a45f54df369d37b9d80a1f86086e9406876072b":{"kind":"REWORK","_number":1,"created":"2023-02-03 06:53:44.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/25/872625/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/25/872625/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/25/872625/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/25/872625/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/25/872625/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/25/872625/1"}}},"commit":{"parents":[{"commit":"c5953f174a3b4a12c970fcd357c740c3fc79906c","subject":"fix keypair showing in magnum service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c5953f174a3b4a12c970fcd357c740c3fc79906c"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-02-03 06:28:14.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-02-03 06:53:18.000000000","tz":480},"subject":"fix: Modify label names to make them consistent","message":"fix: Modify label names to make them consistent\n\nfix \u0027restart policy\u0027 to \u0027exit policy\u0027 in detail page to match the creation process\n\nChange-Id: I4740c377107a4e8687667c3e24578e1f7846aaba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5a45f54df369d37b9d80a1f86086e9406876072b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5a45f54df369d37b9d80a1f86086e9406876072b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[]}
