)]}'
{"id":"openstack%2Fskyline-console~873516","triplet_id":"openstack%2Fskyline-console~master~Ibf5760e88d7772facc4960f0f1f947444dfc227f","project":"openstack/skyline-console","branch":"master","topic":"fix-magnum-ui","attention_set":{},"removed_from_attention_set":{"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2023-02-14 01:50:54.000000000","reason":"\u003cGERRIT_ACCOUNT_28706\u003e replied on the change","reason_account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2023-02-14 03:07:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibf5760e88d7772facc4960f0f1f947444dfc227f","subject":"fix: fix the resize of magnum cluster","status":"MERGED","created":"2023-02-13 10:00:52.000000000","updated":"2023-02-14 03:09:04.000000000","submitted":"2023-02-14 03:07:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873516-fix-magnum-ui","meta_rev_id":"1f7302002aca14f7723e23a4e6baefbe54038e54","_number":873516,"virtual_id_number":873516,"owner":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-14 03:07:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-02-14 01:50:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-02-14 01:50:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-13 10:15:31.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"},{"updated":"2023-02-13 11:26:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"56e77686ab7c7a70d0e39d22d623ed8d54ecc540","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-02-13 10:00:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"77618a780af6ff0d39406fa8315c17bc861a67ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 11:26:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f039b7df7827445e809a977780e3b40f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d67e74fb11504532bbf60e23b87204f6 : SUCCESS in 4m 16s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/ae23d26b964743b59da4bbc10f2f73fe : SUCCESS in 8m 51s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/144874773a99491bb9aafbb3f9178f6e : SUCCESS in 9m 30s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/932bf390b45b46dca76c44614ed94315 : SUCCESS in 6m 30s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/c294bb38e2524ab097fd13c12a6daf24 : FAILURE in 57m 09s","accounts_in_message":[],"_revision_number":1},{"id":"b10b896fa1ee5d372bcc93bcec582e4d5aad9231","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-02-13 13:21:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"429360bf9df0972bb42dd942ddb2b8e61d6ca0bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 14:17:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d6626f0880a445e8a96858513d9a5235\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0130cd5a7d804c96be5489476e389864 : SUCCESS in 4m 46s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/a3da8adab8fd47a29d421b20d9be0331 : SUCCESS in 15m 57s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/c98b8aa165b44847895cfa39628a67da : SUCCESS in 7m 50s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/8485154f36864f8bbe2e5c50f3ec4cf2 : SUCCESS in 21m 15s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/9267b406d619495f8acc8fe9faba1e3f : FAILURE in 46m 25s","accounts_in_message":[],"_revision_number":1},{"id":"30fc12726351334c8cdcfe55702cef891f753d10","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-02-13 14:23:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c4d73b034010635d2ba28d325a85f7967b1b65b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 15:05:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b29baf3e7a94a49a8c5ffb2f95d1dba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d540870fb844a80a895c42dabd40d94 : SUCCESS in 5m 23s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/90949b6ee7744bb49e113d504c8c0df0 : SUCCESS in 9m 03s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/f2d3f99d5e0e4b1fb1f65f341a326290 : SUCCESS in 5m 26s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/0ad89feee72e4a6f8fb5f8efcefe5ad6 : SUCCESS in 5m 47s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/040c8385a7be43a3b0a8191de7cdad32 : SUCCESS in 40m 42s","accounts_in_message":[],"_revision_number":1},{"id":"3e2c06d15cbb9b24057c62106d1ae9ca13d48ec1","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-02-14 01:50:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2378e3f40a9ecb2b4537aa2f45273e2927dcc7a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 01:51:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aeb25915bd17aa81428f25aa87bc159570ece4ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 03:07:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b12e3ee285040c4ac81ed52ad797706\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1234c68372024673858a1655dea962ba : SUCCESS in 4m 42s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/01bb252e07604624a44b4c2fd6c6146a : SUCCESS in 11m 38s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/92d418b7851c4b36aec01071713ee59c : SUCCESS in 7m 14s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/763497501f71409592d512b0c143a15c : SUCCESS in 7m 44s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/94df067f5f824871a8c289dac5d5c4de : SUCCESS in 59m 27s","accounts_in_message":[],"_revision_number":1},{"id":"9052c60990a36674c886475fd4da04c278a5ac24","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 03:07:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1f7302002aca14f7723e23a4e6baefbe54038e54","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 03:09:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b25a106b887456284463df57528f462\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31ad45f42fe045159745b9ed3fd54e81 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"974427c3d8a1652e4f61387bfc121bcae2ea9f32","revisions":{"974427c3d8a1652e4f61387bfc121bcae2ea9f32":{"kind":"REWORK","_number":1,"created":"2023-02-13 10:00:52.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/16/873516/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/16/873516/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/16/873516/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/16/873516/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/16/873516/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/16/873516/1"}}},"commit":{"parents":[{"commit":"d5d14d7637640384476caa85fd659d1f07b3498d","subject":"feature: Support upgrade cluster and show quota info when resize cluster","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d5d14d7637640384476caa85fd659d1f07b3498d"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-02-13 10:00:34.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-02-13 10:00:34.000000000","tz":480},"subject":"fix: fix the resize of magnum cluster","message":"fix: fix the resize of magnum cluster\n\n1. specify the request header `OpenStack-API-Version: container-infra latest` to fix response code 406\n2. fix the nodes_to_remove according to the api ref document\n\nChange-Id: Ibf5760e88d7772facc4960f0f1f947444dfc227f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/974427c3d8a1652e4f61387bfc121bcae2ea9f32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/974427c3d8a1652e4f61387bfc121bcae2ea9f32"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[]}
