)]}'
{"id":"openstack%2Fskyline-console~873844","triplet_id":"openstack%2Fskyline-console~master~I1638b3150059225b582a6e2a14e66b022934033b","project":"openstack/skyline-console","branch":"master","topic":"fix-zun","attention_set":{},"removed_from_attention_set":{"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2023-02-15 08:20:59.000000000","reason":"\u003cGERRIT_ACCOUNT_28706\u003e replied on the change","reason_account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2023-02-15 09:03:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1638b3150059225b582a6e2a14e66b022934033b","subject":"fix: fix the zun-ui","status":"MERGED","created":"2023-02-15 07:01:04.000000000","updated":"2023-02-15 09:04:28.000000000","submitted":"2023-02-15 09:03:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873844-fix-zun","meta_rev_id":"5c3cd904068cc95520d500775925f58565b1fc41","_number":873844,"virtual_id_number":873844,"owner":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-15 09:03:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-02-15 08:20:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-02-15 08:20:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-15 08:00:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-15 08:00:55.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"d0f95e3ce3f55f78a22123b5c30dee32ad15c895","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-02-15 07:01:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b32972c2a17cf1627238caec03333b2b0347d1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 08:00:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6b4bc9719054298a3ff1e77bbcbf68f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/773dc15a047349cdb79683dc4681caaf : SUCCESS in 4m 15s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/4a13b6d471754b9ba4100238ddef8a09 : SUCCESS in 15m 00s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/324ba6d125af44529e8602664cc254d1 : SUCCESS in 5m 32s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/e6555699526d4faeaa1cd3024e545543 : SUCCESS in 12m 48s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/8c8237931f394419a113cc22f8be5c8e : SUCCESS in 56m 48s","accounts_in_message":[],"_revision_number":1},{"id":"8c7f0c2f1fca3ef57f93a0ccd1d311099dc03073","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-02-15 08:20:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5b74b233df5043890ebf2d175d18be63dce87af5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 08:21:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d48ea2de48db28e9fcb03872a3921c6a54d6a0a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 09:03:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97f21a4e599e4d9ab072d1b8544f409d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3e7213fccc449e6b32969af6123e159 : SUCCESS in 4m 46s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/fc9a61bec25a4fb385590cf588e3b472 : SUCCESS in 8m 09s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/7b625573cd00421dbae86f1b5817c6cd : SUCCESS in 5m 11s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/dd37d1c4238346b1a5fa3184068dec79 : SUCCESS in 5m 17s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/d05c95823d24403abe682a3d826bc691 : SUCCESS in 40m 00s","accounts_in_message":[],"_revision_number":1},{"id":"2991afcbe75a03d10d900f799d47797bceb8c983","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 09:03:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5c3cd904068cc95520d500775925f58565b1fc41","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 09:04:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4eba2ac8692141dd9af420f97454bb08\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f403dde3481546078b8baba7f88b984c : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"01ea21d118b0424a8cc3405e213a728883867f9a","revisions":{"01ea21d118b0424a8cc3405e213a728883867f9a":{"kind":"REWORK","_number":1,"created":"2023-02-15 07:01:04.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/44/873844/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/44/873844/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/44/873844/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/44/873844/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/44/873844/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/44/873844/1"}}},"commit":{"parents":[{"commit":"974427c3d8a1652e4f61387bfc121bcae2ea9f32","subject":"fix: fix the resize of magnum cluster","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/974427c3d8a1652e4f61387bfc121bcae2ea9f32"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-02-15 07:00:08.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-02-15 07:00:08.000000000","tz":480},"subject":"fix: fix the zun-ui","message":"fix: fix the zun-ui\n\n1. supplement the task_state of zun containers\n\n2. add search filters to action logs\n\n3. fix the name filter in zun services\n\nChange-Id: I1638b3150059225b582a6e2a14e66b022934033b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/01ea21d118b0424a8cc3405e213a728883867f9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/01ea21d118b0424a8cc3405e213a728883867f9a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[]}
