)]}'
{"id":"openstack%2Fskyline-console~874082","triplet_id":"openstack%2Fskyline-console~master~I78168fc6299c0369a4057b5ed2e01a502adc07c2","project":"openstack/skyline-console","branch":"master","topic":"clean-static","hashtags":[],"change_id":"I78168fc6299c0369a4057b5ed2e01a502adc07c2","subject":"Ignore autogenerated static files","status":"ABANDONED","created":"2023-02-16 10:53:00.000000000","updated":"2023-02-22 11:27:07.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"4c64b0d543f01dab635fc0fed724cf5e50f394cd","_number":874082,"virtual_id_number":874082,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-02-16 12:29:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2023-02-16 12:29:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f56573f8d1ce18e4b9b2af9ea7e1abb0f5913b86","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-02-16 10:53:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e213b2c389815bb7c5888336b704c25bda104bba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 12:29:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0350113d4d6e40f7939df003e278ccdd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afe6311d82424a72895031919a88be13 : SUCCESS in 4m 08s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/3001efe94bcd491eaf13a443f8741f27 : SUCCESS in 9m 56s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/615b163ce4344282938f5675b0b13349 : SUCCESS in 8m 18s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/d5da7503360a4b88900e4b0ad7204ff6 : SUCCESS in 7m 31s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/8a1e0f3a6369496282c51fa56e083c0a : SUCCESS in 1h 18m 46s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/8050dbd8959047fc85128351034c03ac : SUCCESS in 59m 38s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/b6c27ccd0330487cbe38a10b21e07fd7 : SUCCESS in 53m 14s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/1d5ae8bbc6b04ccf96e5f0e793c79615 : SUCCESS in 1h 09m 02s","accounts_in_message":[],"_revision_number":1},{"id":"4c64b0d543f01dab635fc0fed724cf5e50f394cd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-02-22 11:27:07.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"caea788c85aa20388bdf2151560065e5d3c7e5a2","revisions":{"caea788c85aa20388bdf2151560065e5d3c7e5a2":{"kind":"REWORK","_number":1,"created":"2023-02-16 10:53:00.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/82/874082/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/82/874082/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/874082/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/874082/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/874082/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/82/874082/1"}}},"commit":{"parents":[{"commit":"fbc8fde8089490f9054c02e933e07444dd88d4a1","subject":"Merge \"feat: add attributes to the ModalAction\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/fbc8fde8089490f9054c02e933e07444dd88d4a1"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-02-16 10:42:16.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-02-16 10:52:42.000000000","tz":60},"subject":"Ignore autogenerated static files","message":"Ignore autogenerated static files\n\nChange-Id: I78168fc6299c0369a4057b5ed2e01a502adc07c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/caea788c85aa20388bdf2151560065e5d3c7e5a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/caea788c85aa20388bdf2151560065e5d3c7e5a2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
