)]}'
{"id":"openstack%2Fskyline-console~874194","triplet_id":"openstack%2Fskyline-console~master~I1ed2afa24394880c9bdb79c5b75ef9ec6b2ed84a","project":"openstack/skyline-console","branch":"master","topic":"fix-zun","attention_set":{},"removed_from_attention_set":{"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2023-02-17 08:04:50.000000000","reason":"\u003cGERRIT_ACCOUNT_28706\u003e replied on the change","reason_account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2023-02-17 09:21:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1ed2afa24394880c9bdb79c5b75ef9ec6b2ed84a","subject":"fix: add default value and tip when create zun container","status":"MERGED","created":"2023-02-17 07:00:50.000000000","updated":"2023-02-17 09:22:41.000000000","submitted":"2023-02-17 09:21:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874194-fix-zun","meta_rev_id":"3b75a3965dd613b095a6cab07cdaa7c38d9cb47b","_number":874194,"virtual_id_number":874194,"owner":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-17 09:21:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-02-17 08:04:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-02-17 08:04:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-17 08:01:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-17 08:02:32.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"47c8258e61a740dd28b39f087dace86fc8718d8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-02-17 07:00:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"591f15ed9e46a14bd0a872a8f4c41365e0c149e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 08:01:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff3e6610e6184e28bf26055375c690a6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5cf247727f14ba996272040d945c85a : SUCCESS in 4m 19s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/b1dd6a99bbd34ae5b87601adf3dfd2f2 : SUCCESS in 10m 45s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/edc0f94c576448b1a23dc609ae6fa7f2 : SUCCESS in 7m 41s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/cef01f9e323e4f9e8b05180d162db70a : SUCCESS in 7m 26s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/b3d9371c3ac346fe9cb5ab75885f0d69 : SUCCESS in 55m 04s","accounts_in_message":[],"_revision_number":1},{"id":"c59837c6f98c45f0d5789c215761aed96690dffb","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-02-17 08:04:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4decc58fa8145d7fa6035a563b3e35820bfab3f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 08:05:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a8c3c3fa24db11682d7d0e0265296444773a7408","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 09:21:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a6e3349ab5349468fad43a7f92acddf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da75c8f1f5aa403daadf6d3bfc6957c8 : SUCCESS in 4m 10s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/d473cc980c9145da980ac99b6e4077de : SUCCESS in 11m 02s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/8ab8960944864a72bc7b594cf4f0a36c : SUCCESS in 6m 49s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/42a92a9c57fc4bb191999ff7fa017786 : SUCCESS in 6m 54s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/4128c3958d30484a8959a58eab8e9811 : SUCCESS in 55m 42s","accounts_in_message":[],"_revision_number":1},{"id":"f68eb8e0a0827474264cd418a83750a2e9801cfe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 09:21:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3b75a3965dd613b095a6cab07cdaa7c38d9cb47b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 09:22:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51d4091d91474bdba521cafff29d7ff1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa57bb9d464740ccaab0fe4a394e2ad1 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"929842c82453a91f9062ced642d79b1273f539a3","revisions":{"929842c82453a91f9062ced642d79b1273f539a3":{"kind":"REWORK","_number":1,"created":"2023-02-17 07:00:50.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/94/874194/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/94/874194/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/94/874194/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/94/874194/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/94/874194/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/94/874194/1"}}},"commit":{"parents":[{"commit":"635f829217a8ff001709a39a9728b9f87553e161","subject":"fix: fix the rebuild container and ports of zun","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/635f829217a8ff001709a39a9728b9f87553e161"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-02-17 06:58:40.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-02-17 06:58:40.000000000","tz":480},"subject":"fix: add default value and tip when create zun container","message":"fix: add default value and tip when create zun container\n\n1. Add the default value of cpu, memory, hard disk and check quota immediately\n\n2. Modify the hint about ports to be more detailed\n\nChange-Id: I1ed2afa24394880c9bdb79c5b75ef9ec6b2ed84a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/929842c82453a91f9062ced642d79b1273f539a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/929842c82453a91f9062ced642d79b1273f539a3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[]}
