)]}'
{"id":"openstack%2Fskyline-console~874348","triplet_id":"openstack%2Fskyline-console~master~I9e3232dc5828208d3ef08b033d953501fe93fed8","project":"openstack/skyline-console","branch":"master","topic":"fix-network-ip-availability","attention_set":{},"removed_from_attention_set":{"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2023-03-01 10:56:34.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-03-01 07:55:39.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2023-03-01 07:57:07.000000000","reason":"\u003cGERRIT_ACCOUNT_33689\u003e replied on the change","reason_account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"}}},"hashtags":[],"change_id":"I9e3232dc5828208d3ef08b033d953501fe93fed8","subject":"fix: fix the network IP availability showing","status":"MERGED","created":"2023-02-18 14:17:09.000000000","updated":"2023-03-01 10:57:39.000000000","submitted":"2023-03-01 10:56:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874348-fix-network-ip-availability","meta_rev_id":"c522608b583d0c2a7eeb702ddb4bf7da0809d02c","_number":874348,"virtual_id_number":874348,"owner":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-01 10:56:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"recommended":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2023-03-01 07:55:39.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":2,"date":"2023-03-01 09:45:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":1,"date":"2023-03-01 09:45:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-18 15:35:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-01 02:26:44.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2023-03-01 07:57:07.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"cc8c844653ec6b22d45f33138af12ec800803409","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-02-18 14:17:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"368db6681df11b3d2fe212a36cc3457660c5c938","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-02-18 14:41:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9316618a0a51bf98f58a3c2bbd5df4bd358142b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-18 15:35:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a8066a1a352b4510aaf1825fb30e6765\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b09060da46f4c27aa7a846771efcbfb : SUCCESS in 4m 04s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/8f04e0dc939448ca8f552bf6b6c62224 : SUCCESS in 11m 02s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/4a8ca7ad4d8a4d98881be4e669aef814 : SUCCESS in 4m 29s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/59f27576076c4f479e4c2f4cae311b58 : SUCCESS in 7m 17s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/56c2e63ed34e439d8dbae0ef0199198b : FAILURE in 52m 30s","accounts_in_message":[],"_revision_number":2},{"id":"2946c03d94eb5a6d6bc42dbd1a9b19fed76c74c9","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-02-19 14:25:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7af1137bdaee82630c78704630bd34d67049e4e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-19 15:34:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b50004b15e9d4bb880b67fc61f2364ab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/254d8f8a8f81494cba20bd51153ebaf8 : SUCCESS in 4m 38s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/25950f168c494f92a62eb294ba00d284 : SUCCESS in 9m 59s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/803ca44022274d6eae1662166be2ac08 : SUCCESS in 7m 29s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/1e8cff7bdbec414ebb52b86d3f527645 : SUCCESS in 7m 27s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/620b972fe62c44b6b3c443f75cdd4612 : SUCCESS in 1h 01m 52s","accounts_in_message":[],"_revision_number":2},{"id":"8bf07d0b966c576b16c555890f3d0e68931b1bc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-03-01 06:30:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"81304fc3efa3a6035bd16875201978521e4457ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-03-01 06:37:02.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e72c7c736c79bc71ba0dc87f15414c364b42bcbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 07:47:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c187e7c6ca85455dbb44b11b6fede5e8\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7bd815e30f264bd0bf3b981384193daa : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f4bc7e68ad94b21a338f3d2453a129f : SUCCESS in 3m 59s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/d5289adf2e19436691b8ae294b8a0eb4 : SUCCESS in 7m 09s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/575b19840b804a818ab2dec3eff0c8ad : SUCCESS in 6m 47s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/b6b307c18ac847a2ab6299ee58d3f78f : SUCCESS in 5m 55s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/19809d2870bc4ba88414abb2a510dc11 : SUCCESS in 1h 04m 39s","accounts_in_message":[],"_revision_number":4},{"id":"94812f607cde11bc2ee3432bd6d379429b178393","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-03-01 07:55:39.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"4db9958e25b82d0242f48837d66d73cc19878ad6","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-03-01 09:45:58.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"1ce403cda36a0a43b2cd8fe24e2219bbb182667b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 09:46:35.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a458188acdb275b3be39e13bc94cd18be16ae000","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 10:56:34.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fcc5a7960ae5486a8f52b05080d9fbad\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a99a4ce2ec948aaadb460f4444971a8 : SUCCESS in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f71a823bf6444d08469aa24fb7631b2 : SUCCESS in 4m 07s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/022d9494d01247818755f669a2fe83b9 : SUCCESS in 10m 55s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/e3517543f6984c9abfa0618b69e7313f : SUCCESS in 7m 04s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/d321d317133f463dbaed025baf3bef43 : SUCCESS in 7m 43s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/2006393b3c1c420fb2b57099ea1588c2 : SUCCESS in 1h 04m 57s","accounts_in_message":[],"_revision_number":4},{"id":"d1bfea0b8348a4cf532a99b64b7a83b8dbb7982b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 10:56:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"c522608b583d0c2a7eeb702ddb4bf7da0809d02c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 10:57:39.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68fd7894e88847dba456e91a3527e9a8\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ac7e479dc6846b691bdceae44a63b7f : SUCCESS in 37s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8beebfdf3c341ad87ac364dce9b7185 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"520504e26d177397c5b82eaf9842d259e3744ca5","revisions":{"225e0de1b4ecaa292dd62cf3713a1d6568460920":{"kind":"REWORK","_number":1,"created":"2023-02-18 14:17:09.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/48/874348/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/48/874348/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/874348/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/874348/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/874348/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/48/874348/1"}}},"commit":{"parents":[{"commit":"929842c82453a91f9062ced642d79b1273f539a3","subject":"fix: add default value and tip when create zun container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/929842c82453a91f9062ced642d79b1273f539a3"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-02-18 14:15:52.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-02-18 14:15:52.000000000","tz":480},"subject":"fix: fix the network IP availability showing","message":"fix: fix the network IP availability showing\n\nBy default policy configuration, only administrative users can retrieve IP availability. Otherwise, Not Found (404) will be returned\n\nCloses-Bug: #2003303\nChange-Id: I9e3232dc5828208d3ef08b033d953501fe93fed8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/225e0de1b4ecaa292dd62cf3713a1d6568460920"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/225e0de1b4ecaa292dd62cf3713a1d6568460920"}]},"branch":"refs/heads/master"},"1baa43d14905099d590b5928e40d2d2c270e5434":{"kind":"REWORK","_number":2,"created":"2023-02-18 14:41:20.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/48/874348/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/48/874348/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/874348/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/874348/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/874348/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/48/874348/2"}}},"commit":{"parents":[{"commit":"929842c82453a91f9062ced642d79b1273f539a3","subject":"fix: add default value and tip when create zun container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/929842c82453a91f9062ced642d79b1273f539a3"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-02-18 14:15:52.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-02-18 14:40:41.000000000","tz":480},"subject":"fix: fix the network IP availability showing","message":"fix: fix the network IP availability showing\n\nBy default policy configuration, only administrative users can retrieve IP availability. Otherwise, Not Found (404) will be returned\n\nCloses-Bug: #2003303\nChange-Id: I9e3232dc5828208d3ef08b033d953501fe93fed8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1baa43d14905099d590b5928e40d2d2c270e5434"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1baa43d14905099d590b5928e40d2d2c270e5434"}]},"branch":"refs/heads/master"},"744f7fa0a82671713bd4571e37d58e8bb7171b2f":{"kind":"REWORK","_number":3,"created":"2023-03-01 06:30:27.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/48/874348/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/48/874348/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/874348/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/874348/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/874348/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/48/874348/3"}}},"commit":{"parents":[{"commit":"0a192be466c1fe0ac0ca75f5937b23d56c83fb01","subject":"feature: Support attach network and detach network for zun container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0a192be466c1fe0ac0ca75f5937b23d56c83fb01"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-02-18 14:15:52.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-03-01 06:28:31.000000000","tz":480},"subject":"fix: fix the network IP availability showing","message":"fix: fix the network IP availability showing\n\nBy default policy configuration, only administrative users can retrieve IP availability. Otherwise, Not Found (404) will be returned\n\nCloses-Bug: #2003303\nChange-Id: I9e3232dc5828208d3ef08b033d953501fe93fed8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/744f7fa0a82671713bd4571e37d58e8bb7171b2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/744f7fa0a82671713bd4571e37d58e8bb7171b2f"}]},"branch":"refs/heads/master"},"520504e26d177397c5b82eaf9842d259e3744ca5":{"kind":"REWORK","_number":4,"created":"2023-03-01 06:37:02.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/48/874348/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/48/874348/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/874348/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/874348/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/874348/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/48/874348/4"}}},"commit":{"parents":[{"commit":"0a192be466c1fe0ac0ca75f5937b23d56c83fb01","subject":"feature: Support attach network and detach network for zun container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0a192be466c1fe0ac0ca75f5937b23d56c83fb01"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-02-18 14:15:52.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-03-01 06:36:30.000000000","tz":480},"subject":"fix: fix the network IP availability showing","message":"fix: fix the network IP availability showing\n\nBy default policy configuration, only administrative users can retrieve IP availability. Otherwise, Not Found (404) will be returned\n\nCloses-Bug: #2003303\nChange-Id: I9e3232dc5828208d3ef08b033d953501fe93fed8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/520504e26d177397c5b82eaf9842d259e3744ca5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/520504e26d177397c5b82eaf9842d259e3744ca5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[]}
