)]}'
{"id":"openstack%2Fskyline-console~875617","triplet_id":"openstack%2Fskyline-console~master~I0cfe22e22b1f6ba3ce525a24ca1f95e5fcb098ec","project":"openstack/skyline-console","branch":"master","topic":"fix-zun-ui","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-02-28 09:10:03.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2023-02-28 10:53:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0cfe22e22b1f6ba3ce525a24ca1f95e5fcb098ec","subject":"feature: Support attach network and detach network for zun container","status":"MERGED","created":"2023-02-28 07:19:20.000000000","updated":"2023-02-28 10:54:23.000000000","submitted":"2023-02-28 10:53:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875617-fix-zun-ui","meta_rev_id":"55a55a6a6f1de101bf9d8674f3f2f7866fb9e2f5","_number":875617,"virtual_id_number":875617,"owner":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-28 10:53:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2023-02-28 09:10:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2023-02-28 09:10:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-28 08:54:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-28 09:02:57.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"7f8b65149a27580f8381183fb2fe1d9df308294b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-02-28 07:19:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"01359c980673cf999ac7b396e093da1e9a14958a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 08:54:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec32843ce8c946f4aa542ad26ab88834\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d07a583ec59c4ef69bf84db552dd63eb : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f5b10af82514af794e22cadf6c57530 : SUCCESS in 5m 15s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/ecbc294c3392476e95af711d71da89f6 : SUCCESS in 8m 59s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/4fec8e2bf596461d820a93bbcefb992f : SUCCESS in 8m 34s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/ed8116e1ce574a25835db8391cba667e : SUCCESS in 7m 02s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/3f0c62ea41fc4da397deb8a04e011d26 : SUCCESS in 1h 15m 07s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/6e7a0b1e7ca14610b6882124b04c3d4c : SUCCESS in 57m 28s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/14d7138836b74ee38976329d8279b6d1 : SUCCESS in 49m 53s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/0e3cb27b34684363938956166f16e997 : SUCCESS in 46m 59s","accounts_in_message":[],"_revision_number":1},{"id":"87dfa83da6d7ba06a0f96634aed9a75e08ca0f84","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-02-28 09:10:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0d8de9906c847347632e69d9be53a4fc97a8492f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 09:10:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7466321b7963856abf0c9cb729cc6638bce3f5dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 10:53:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3f275a3bbc64c2ea5ca5ff959cc511e\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21a0dbccf772464d909d818966e6424e : SUCCESS in 2m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af69ed4b3c324ecfbd4bef1e76b144ce : SUCCESS in 4m 26s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/9f01549751774d479941d39ab473fb3f : SUCCESS in 9m 58s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/e63fe399d09b44dbac3f66391110c192 : SUCCESS in 8m 30s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/449f51a7d1f442c28425fe7e0fc4a53d : SUCCESS in 7m 20s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/5cddf941e7234b2dacb0e96208f744a7 : SUCCESS in 1h 21m 46s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/cd3112743fde4e6eb8e65a66c801cf00 : SUCCESS in 1h 07m 05s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/67c10143cec848a7ba601a0c0db80f23 : SUCCESS in 57m 23s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/8e54550805ec4bfc8a83f06f852a6fdd : SUCCESS in 1h 02m 16s","accounts_in_message":[],"_revision_number":1},{"id":"613cb04cbac120cab17310a82d94d5a3ce1a5b44","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 10:53:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"55a55a6a6f1de101bf9d8674f3f2f7866fb9e2f5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 10:54:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfaa9a96d3ff45c19c345b4018449ee1\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/02b426f37fd1483a894360fe0879e02a : SUCCESS in 38s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9d5560fb143479e8ca87dd1c9ad0786 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0a192be466c1fe0ac0ca75f5937b23d56c83fb01","revisions":{"0a192be466c1fe0ac0ca75f5937b23d56c83fb01":{"kind":"REWORK","_number":1,"created":"2023-02-28 07:19:20.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/17/875617/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/17/875617/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/17/875617/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/17/875617/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/17/875617/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/17/875617/1"}}},"commit":{"parents":[{"commit":"929842c82453a91f9062ced642d79b1273f539a3","subject":"fix: add default value and tip when create zun container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/929842c82453a91f9062ced642d79b1273f539a3"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-02-28 07:16:34.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-02-28 07:16:34.000000000","tz":480},"subject":"feature: Support attach network and detach network for zun container","message":"feature: Support attach network and detach network for zun container\n\n1. Add attach network action for zun container\n\n2. Add detach network action for zun container\n\nChange-Id: I0cfe22e22b1f6ba3ce525a24ca1f95e5fcb098ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0a192be466c1fe0ac0ca75f5937b23d56c83fb01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0a192be466c1fe0ac0ca75f5937b23d56c83fb01"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[]}
