)]}'
{"id":"openstack%2Fskyline-console~875876","triplet_id":"openstack%2Fskyline-console~stable%2F2023.1~Ie569bfb335f4972dbda500ce142cf280566afcf2","project":"openstack/skyline-console","branch":"stable/2023.1","topic":"create-2023.1","hashtags":[],"change_id":"Ie569bfb335f4972dbda500ce142cf280566afcf2","subject":"Update .gitreview for stable/2023.1","status":"MERGED","created":"2023-03-01 11:13:32.000000000","updated":"2023-03-24 16:17:06.000000000","submitted":"2023-03-24 16:15:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875876-create-2023.1","meta_rev_id":"27287582edefea6ee0339e5c8b7164af57238662","_number":875876,"virtual_id_number":875876,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-24 16:15:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-03-24 14:21:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-03-24 14:21:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-01 11:50:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-24 14:21:35.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"57dae479acef9f6bba5f91f1c28cce7a70f01a7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2023-03-01 11:13:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9da205fff2434ef023f95c40c067d16c67cc0a9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 11:50:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be3515005c21498fa18b72c602f50505\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2e0764d6eb74955aeb4fca5ef7d4cb4 : SUCCESS in 4m 19s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/9b1aec0002c940fba9507b7ebdb1f5fd : SUCCESS in 10m 13s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/bdfbf1635cb54623a0f49a8aa43ed062 : SUCCESS in 6m 58s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/a8f91edab35f44fd80273c7b497ebe3d : SUCCESS in 8m 34s","accounts_in_message":[],"_revision_number":1},{"id":"e483e12e18d4c6513d46f9e89550d55b89636b87","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-03-24 14:21:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2b9233426c8c517cecf41ac2e49c3bbc69081577","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 14:22:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dd236c615f6ff7c35d3d20d614b9c09c0ff9d28e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 16:15:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02ad28aff9234a5eabf94c9bf7602488\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d17b1a5d4be430f9a5f89127693ae05 : SUCCESS in 5m 26s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/20c9f9c12adb4c71b5e9d3e93d2c1c94 : SUCCESS in 8m 09s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/3963044af7c646b9a047240472855e04 : SUCCESS in 5m 42s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/83725df328574a99bb9c257b552fdba4 : SUCCESS in 5m 49s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/dc765419ef7448b385b994ef3111b374 : SUCCESS in 1h 18m 13s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/74d2374f75e448a7aadb273f1ee654e4 : SUCCESS in 53m 19s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/2720d2f9a5f44373a0809e6b0af06342 : SUCCESS in 45m 00s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/0c8e47a2a6124b9295491c95284f8673 : SUCCESS in 53m 03s","accounts_in_message":[],"_revision_number":1},{"id":"31a7f7e12fba7f52d7353f875ad7976364af8357","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 16:15:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"27287582edefea6ee0339e5c8b7164af57238662","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 16:17:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/061d1141aec943718c71fac6bf46c68f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9e2314bda2d4664b18a193b31aba0d0 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b6c6d6006c53416e29d4bf9e8a6cc9310ef5bf53","revisions":{"b6c6d6006c53416e29d4bf9e8a6cc9310ef5bf53":{"kind":"REWORK","_number":1,"created":"2023-03-01 11:13:32.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/76/875876/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/76/875876/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/76/875876/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/76/875876/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/76/875876/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/76/875876/1"}}},"commit":{"parents":[{"commit":"0a192be466c1fe0ac0ca75f5937b23d56c83fb01","subject":"feature: Support attach network and detach network for zun container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0a192be466c1fe0ac0ca75f5937b23d56c83fb01"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-03-01 11:13:31.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-03-01 11:13:31.000000000","tz":0},"subject":"Update .gitreview for stable/2023.1","message":"Update .gitreview for stable/2023.1\n\nChange-Id: Ie569bfb335f4972dbda500ce142cf280566afcf2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b6c6d6006c53416e29d4bf9e8a6cc9310ef5bf53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b6c6d6006c53416e29d4bf9e8a6cc9310ef5bf53"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
