)]}'
{"id":"openstack%2Fskyline-console~876493","triplet_id":"openstack%2Fskyline-console~master~Ic9b9c622d2dcf268ac1a366719e52c9169c30c82","project":"openstack/skyline-console","branch":"master","hashtags":[],"change_id":"Ic9b9c622d2dcf268ac1a366719e52c9169c30c82","subject":"docs: Remove invalid README.rst","status":"MERGED","created":"2023-03-06 07:34:54.000000000","updated":"2023-03-06 08:04:56.000000000","submitted":"2023-03-06 08:03:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876493","meta_rev_id":"493f1f0d4fe79eb77a27a77b71124c5b91ffb24d","_number":876493,"virtual_id_number":876493,"owner":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-06 08:03:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-03-06 07:37:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-03-06 07:37:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-06 07:49:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d6b781eb2d7479f7a02b3b35d38098763148daf5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-03-06 07:34:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba99b38ed3bcbd8a9c7296e72b00c09b98d75896","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-03-06 07:37:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"02701f510d6d82ee2ba337ebf7582c728511293e","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-03-06 07:37:31.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"efbaebcf7f070e8c2faa3d774b263d12b8dde107","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 07:49:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4666a059022c41fbae127384f454bff4\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d23fab52cd044ddb80d3f93be2c78918 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96dd5c98a977469d87e86eca8734eb42 : SUCCESS in 4m 27s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/32db22e26e0f4c09bdf75c7b2af5b7d8 : SUCCESS in 9m 27s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/c76de3a7be7f430c867c83e07dda5ef3 : SUCCESS in 6m 56s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/b4090802dd6541ad8628b9e53dd569d3 : SUCCESS in 6m 48s","accounts_in_message":[],"_revision_number":1},{"id":"2e0f5642addae3250ee1f14243d8ab45d5b10040","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 07:49:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ebbbf7d7e6eecd62ee8aa85ce95ce895a6ad5825","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 08:03:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d25452abefc94eaab824c03ae056e080\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e8ffc909f8d4a49a6f3da0afcb61e1e : SUCCESS in 2m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b3ff7c6b3084db8b6a1894149a7dccd : SUCCESS in 4m 39s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/a02a918b2ecf49de865f4d390a4f8281 : SUCCESS in 9m 31s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/470335ab3e4145c9bea26f3434a82f4e : SUCCESS in 6m 42s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/2d00cff1bf0940d88b7e0cab51a0d82b : SUCCESS in 7m 50s","accounts_in_message":[],"_revision_number":1},{"id":"6fbc64cec7e8ba77ad442b91592c975cd173c4dc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 08:03:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"493f1f0d4fe79eb77a27a77b71124c5b91ffb24d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 08:04:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e6f029b094c4462b006bc997cd5132e\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44a1ddcb720b4b0592c8cf83aa889f99 : SUCCESS in 35s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb3295f4ae664a7fb1458a411170fc6a : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4965609ca9796811110e24eb76b330d816e9cd46","revisions":{"4965609ca9796811110e24eb76b330d816e9cd46":{"kind":"REWORK","_number":1,"created":"2023-03-06 07:34:54.000000000","uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/93/876493/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/93/876493/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/876493/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/876493/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/876493/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/93/876493/1"}}},"commit":{"parents":[{"commit":"520504e26d177397c5b82eaf9842d259e3744ca5","subject":"fix: fix the network IP availability showing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/520504e26d177397c5b82eaf9842d259e3744ca5"}]}],"author":{"name":"Boxiang Zhu","email":"zhu.boxiang@99cloud.net","date":"2023-03-06 07:33:56.000000000","tz":480},"committer":{"name":"Boxiang Zhu","email":"zhu.boxiang@99cloud.net","date":"2023-03-06 07:33:56.000000000","tz":480},"subject":"docs: Remove invalid README.rst","message":"docs: Remove invalid README.rst\n\nThe README.rst for releasenotes is wrong and invalid. So remove it.\n\nChange-Id: Ic9b9c622d2dcf268ac1a366719e52c9169c30c82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4965609ca9796811110e24eb76b330d816e9cd46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4965609ca9796811110e24eb76b330d816e9cd46"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[]}
