)]}'
{"id":"openstack%2Fskyline-console~877090","triplet_id":"openstack%2Fskyline-console~master~I062210ea8cf05aea2d017acddb62d0e1d308a24b","project":"openstack/skyline-console","branch":"master","hashtags":[],"change_id":"I062210ea8cf05aea2d017acddb62d0e1d308a24b","subject":"feat: add renderHeader in TabList","status":"ABANDONED","created":"2023-03-10 10:31:05.000000000","updated":"2023-03-15 02:01:59.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8f44afa6adcb7e33bc997e865b1128012a3617a2","_number":877090,"virtual_id_number":877090,"owner":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-03-14 09:49:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-10 11:52:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6fb34fc13fbf3b935fa6a0288b2b7a8d1100c78a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2023-03-10 10:31:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"95c0e4f7201cdd266de8924159624b8d7161fc96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 11:52:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e72378a59044475958323b1a91b2fcc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8acc020b754466e9b6b06df805b0159 : SUCCESS in 4m 56s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/5c17fefc7c0741f7b9d5c7435e8c0fe1 : SUCCESS in 7m 52s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/893bc7c0f396431f8d1c175adf058515 : SUCCESS in 5m 25s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/c07a17f49b5e417b9f1ba917aa17f859 : SUCCESS in 5m 08s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/52d8a1b5f34d42a8a4e17a6cd564e312 : SUCCESS in 1h 18m 45s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/3315a8689630426e9601086be0ab12e6 : SUCCESS in 1h 12m 05s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/bc67206c9d92428eaa31a897d8019abc : SUCCESS in 50m 15s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/797c96d715e24566aa932d4de7f99da9 : SUCCESS in 49m 01s","accounts_in_message":[],"_revision_number":1},{"id":"dc4696fcab7efccfcb2cef5bd253cb4b92b8121a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2023-03-13 09:40:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0161f86ea8e505c6882743a1bd32b812f389ab65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 11:01:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/969d6822257d4f3bbd48ace398811bfd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a7b617e50c34b8f827b1ee369ada127 : SUCCESS in 4m 05s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/78b047c48669425ca17746f9b2709676 : SUCCESS in 7m 51s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/6c248c68364d42089ea6ca1ccc4a3797 : SUCCESS in 6m 07s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/f20da1ab8475405ba568967be2df7957 : SUCCESS in 6m 23s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/2d22626a323d43e1be496ec28083bc94 : SUCCESS in 1h 18m 36s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/50cfd81a246b4e87b2962cd2dcc8401e : SUCCESS in 51m 41s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/bcba35c94050475ab3931561ef8265cb : SUCCESS in 41m 22s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/2ee0a37d1f6142669e1a065e958deb08 : SUCCESS in 47m 04s","accounts_in_message":[],"_revision_number":2},{"id":"b836ca91277ec1427b9729d88a6ff76c996b14d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2023-03-14 08:25:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d6e283d9dd95fcd2c14113d2a6929ad2385e1b1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 09:49:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94eda17c929b492ab7558eb73d854a88\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b05e50f7835454db89f845920dd7485 : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8d2c3cf2af94b58b38aae2bf15c5957 : SUCCESS in 5m 12s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/02b8de013308439b9d8e22d68dcc4d80 : SUCCESS in 7m 16s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/32396148873a4c3f834a2703db3f9e33 : SUCCESS in 7m 36s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/637fea553a6a4dbd873feb6f93a35d2e : SUCCESS in 7m 56s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/d63bba23a25f40c5a125de4ae85d09f3 : SUCCESS in 1h 14m 54s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/c9d46d3bb4b44e649a2e348b59641da6 : SUCCESS in 1h 08m 02s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/7c5e1f92b82843ac9cb5618507a05674 : SUCCESS in 1h 00m 13s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/ee9d95ea2f76438888b1e68af5759ef4 : SUCCESS in 48m 22s","accounts_in_message":[],"_revision_number":3},{"id":"8f44afa6adcb7e33bc997e865b1128012a3617a2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2023-03-15 02:01:59.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"45cc19c7b0724532b79fd114d9bc2c89a7fe4cdc","revisions":{"f6570937c2e1150b9760fd3745d5e05978aff96e":{"kind":"REWORK","_number":1,"created":"2023-03-10 10:31:05.000000000","uploader":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"ref":"refs/changes/90/877090/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/90/877090/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/90/877090/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/90/877090/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/90/877090/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/90/877090/1"}}},"commit":{"parents":[{"commit":"4965609ca9796811110e24eb76b330d816e9cd46","subject":"docs: Remove invalid README.rst","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4965609ca9796811110e24eb76b330d816e9cd46"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2023-03-10 10:28:41.000000000","tz":480},"committer":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2023-03-10 10:28:41.000000000","tz":480},"subject":"feat: add renderHeader in TabList and Topology","message":"feat: add renderHeader in TabList and Topology\n\nAdd renderHeader to support render header for show module instruction\n\nChange-Id: I062210ea8cf05aea2d017acddb62d0e1d308a24b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f6570937c2e1150b9760fd3745d5e05978aff96e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f6570937c2e1150b9760fd3745d5e05978aff96e"}]},"branch":"refs/heads/master"},"6891649c1a8cb3647eb81d4a8604f74a5601be9f":{"kind":"REWORK","_number":2,"created":"2023-03-13 09:40:50.000000000","uploader":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"ref":"refs/changes/90/877090/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/90/877090/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/90/877090/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/90/877090/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/90/877090/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/90/877090/2"}}},"commit":{"parents":[{"commit":"4965609ca9796811110e24eb76b330d816e9cd46","subject":"docs: Remove invalid README.rst","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4965609ca9796811110e24eb76b330d816e9cd46"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2023-03-10 10:28:41.000000000","tz":480},"committer":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2023-03-13 09:39:10.000000000","tz":480},"subject":"feat: add renderHeader in TabList and Topology","message":"feat: add renderHeader in TabList and Topology\n\nAdd renderHeader to support render header for show module instruction\n\nChange-Id: I062210ea8cf05aea2d017acddb62d0e1d308a24b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6891649c1a8cb3647eb81d4a8604f74a5601be9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6891649c1a8cb3647eb81d4a8604f74a5601be9f"}]},"branch":"refs/heads/master"},"45cc19c7b0724532b79fd114d9bc2c89a7fe4cdc":{"kind":"REWORK","_number":3,"created":"2023-03-14 08:25:25.000000000","uploader":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"ref":"refs/changes/90/877090/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/90/877090/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/90/877090/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/90/877090/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/90/877090/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/90/877090/3"}}},"commit":{"parents":[{"commit":"decd2bfec72f508d9d19085aa94aede567808d21","subject":"fix: Fix the label of node count in magnum cluster","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/decd2bfec72f508d9d19085aa94aede567808d21"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2023-03-10 10:28:41.000000000","tz":480},"committer":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2023-03-14 08:24:52.000000000","tz":480},"subject":"feat: add renderHeader in TabList","message":"feat: add renderHeader in TabList\n\nAdd renderHeader to support render header for show module instruction\n\nChange-Id: I062210ea8cf05aea2d017acddb62d0e1d308a24b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/45cc19c7b0724532b79fd114d9bc2c89a7fe4cdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/45cc19c7b0724532b79fd114d9bc2c89a7fe4cdc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
