)]}'
{"id":"openstack%2Fskyline-console~877222","triplet_id":"openstack%2Fskyline-console~master~I5ee4cc8ab3ec084b31d46a9623c729189522c01b","project":"openstack/skyline-console","branch":"master","topic":"fix-magnum-ui","attention_set":{},"removed_from_attention_set":{"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2023-03-13 09:07:46.000000000","reason":"\u003cGERRIT_ACCOUNT_28706\u003e replied on the change","reason_account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-03-13 09:01:28.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2023-03-13 09:49:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5ee4cc8ab3ec084b31d46a9623c729189522c01b","subject":"fix: Fix the label of node count in magnum cluster","status":"MERGED","created":"2023-03-13 07:36:26.000000000","updated":"2023-03-13 09:50:47.000000000","submitted":"2023-03-13 09:49:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877222-fix-magnum-ui","meta_rev_id":"95d14c22827fb88b6a30b1ba5add549a44af73c5","_number":877222,"virtual_id_number":877222,"owner":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-13 09:49:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2023-03-13 09:01:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":2,"date":"2023-03-13 09:07:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":1,"date":"2023-03-13 09:07:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-13 08:42:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-13 09:00:22.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"},{"updated":"2023-03-13 09:00:22.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"935c25ff16b0cb33d81887799b62658d1d974cbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-03-13 07:36:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"33ebe7a316fffb05f2e5085bdf6a3929a6c063d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-03-13 07:39:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"135cf4c009172f95c1818e9a69963d1c445d355a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 08:42:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/643adf8f38474f47ab37e52a0aa89d21\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3d31e9f41914e50ae16ed172aec91c7 : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bad3fe34b3534038a7bb9cacd05d5142 : SUCCESS in 4m 19s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/e42dc11614604f03bcdd3bfd79954533 : SUCCESS in 10m 29s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/16c986720a654c6a9a3bc11a40f190d4 : SUCCESS in 6m 48s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/8d42077d9f1e436d96833016304ac17e : SUCCESS in 6m 17s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/3459b2cf84ab4baf9802957e53aeb148 : SUCCESS in 57m 17s","accounts_in_message":[],"_revision_number":2},{"id":"feb9d945fd8cf9fd53d067aea55d207a41518799","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-03-13 09:01:28.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b2f12bcceca9db65b22c04b3772c17f87114c98f","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-03-13 09:07:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9ed803ce45ec936cc130700492b237355cd18fe6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 09:08:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7d32d6e57ea47a78bdec9d851ec90d8edf40022b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 09:49:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea69c9a014244bf08dffd9cf714d289c\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9481120167604af8897644e526e9f69b : SUCCESS in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c645215b24c4420a9c67a5181f12a0f : SUCCESS in 4m 23s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/10783e02b60f431881f82a6fd9777fb4 : SUCCESS in 8m 40s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/56214ff884804f05a700e097d54635ab : SUCCESS in 6m 00s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/d9b32ded5eae4737b21e90cdc47e9781 : SUCCESS in 6m 17s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/eeea4974b7c0417c884ebf8954a1888d : SUCCESS in 40m 25s","accounts_in_message":[],"_revision_number":2},{"id":"30f254ebe720113a8b34ccdf3f398ef939e1ce44","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 09:49:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"95d14c22827fb88b6a30b1ba5add549a44af73c5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 09:50:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbec880a076c4ceba8d083f2ab468bcb\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ac1f319f4d9473d9c6ff46292283c15 : SUCCESS in 40s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/662b1c7dba264d17a410703669b6b7e6 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"decd2bfec72f508d9d19085aa94aede567808d21","revisions":{"5d95a0296fcc1b268a8e2a540e2ca5c477ea3185":{"kind":"REWORK","_number":1,"created":"2023-03-13 07:36:26.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/22/877222/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/22/877222/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/22/877222/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/22/877222/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/22/877222/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/22/877222/1"}}},"commit":{"parents":[{"commit":"4965609ca9796811110e24eb76b330d816e9cd46","subject":"docs: Remove invalid README.rst","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4965609ca9796811110e24eb76b330d816e9cd46"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-03-13 07:35:23.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-03-13 07:35:44.000000000","tz":480},"subject":"fix: Fix the label of node count in magnum cluster","message":"fix: Fix the label of node count in magnum cluster\n\nUnify the name of the label between detail page and creating step\n\nChange-Id: I5ee4cc8ab3ec084b31d46a9623c729189522c01b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5d95a0296fcc1b268a8e2a540e2ca5c477ea3185"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5d95a0296fcc1b268a8e2a540e2ca5c477ea3185"}]},"branch":"refs/heads/master"},"decd2bfec72f508d9d19085aa94aede567808d21":{"kind":"REWORK","_number":2,"created":"2023-03-13 07:39:49.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/22/877222/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/22/877222/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/22/877222/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/22/877222/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/22/877222/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/22/877222/2"}}},"commit":{"parents":[{"commit":"4965609ca9796811110e24eb76b330d816e9cd46","subject":"docs: Remove invalid README.rst","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4965609ca9796811110e24eb76b330d816e9cd46"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-03-13 07:35:23.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-03-13 07:39:27.000000000","tz":480},"subject":"fix: Fix the label of node count in magnum cluster","message":"fix: Fix the label of node count in magnum cluster\n\nUnify the name of the label between detail page and creating step\n\nChange-Id: I5ee4cc8ab3ec084b31d46a9623c729189522c01b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/decd2bfec72f508d9d19085aa94aede567808d21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/decd2bfec72f508d9d19085aa94aede567808d21"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
