)]}'
{"id":"openstack%2Fskyline-console~878732","triplet_id":"openstack%2Fskyline-console~master~I3e6d3e8aebec4584f863282082a11b36d7cee193","project":"openstack/skyline-console","branch":"master","topic":"i18n","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-03-28 06:48:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3e6d3e8aebec4584f863282082a11b36d7cee193","subject":"feat: update zh-cn to zh-hans","status":"MERGED","created":"2023-03-28 02:16:49.000000000","updated":"2023-03-28 06:49:49.000000000","submitted":"2023-03-28 06:48:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878732-i18n","meta_rev_id":"2197f15daa55f8f7a770be2810216f3d6c79ec99","_number":878732,"virtual_id_number":878732,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-28 06:48:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-03-28 05:13:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":1,"date":"2023-03-28 04:47:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-03-28 05:13:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-28 03:56:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-28 04:47:04.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"state":"REVIEWER"},{"updated":"2023-03-28 05:13:50.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"7be25fd7aa0121194eb567bfd4b85f1e7ceb62f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-03-28 02:16:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e60ea79921b0d9fca32be795b2244b378903e7dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 03:56:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac36216fa8904f61a32b17bb15d49f98\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ec8edc869794498a6263a26643689a8 : SUCCESS in 5m 00s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/7c4c5577be5d44bf9afb3acc2a66b90e : SUCCESS in 9m 24s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/ab57308bc2294df1b6a0f5acfe09fc34 : SUCCESS in 5m 52s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/cb879bc0ed8345bd8fcb027e36734394 : SUCCESS in 5m 54s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/acc1fb79adbd492f98a46aaaa88a180f : SUCCESS in 1h 11m 21s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/38ca295d45fa442194a584806d5fa937 : SUCCESS in 1h 06m 03s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/4ebac1479c42430db3350dd2ad13a560 : SUCCESS in 57m 08s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/2c24656b42924de5932d3847fd172442 : SUCCESS in 1h 01m 19s","accounts_in_message":[],"_revision_number":1},{"id":"966c8d6918dd4e4013d90ecd048a69fe9580a9fc","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-03-28 04:47:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ed744256a53409da0a91e96ea32e5b3faf027f64","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-03-28 05:13:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5d05c22e3f37e214e7773de9533ae68d3411c64a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 05:14:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e9fac1010cda23e9fea7f2c22f033db32e8b9ea1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 06:48:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbfad2a8d97f419b9523ced2e63886de\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ec2450316e04187ba1ac712a08118c1 : SUCCESS in 4m 56s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/17fb30acf7b8485ab86dd30830e6c288 : SUCCESS in 10m 01s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/ad411c40e23e4c14874128ab3e619fc9 : SUCCESS in 6m 59s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/d47f49c87d274a72ac84d38cc66e26fb : SUCCESS in 7m 09s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/af3c6390d8fa42ef8e8de631bd04a142 : SUCCESS in 1h 27m 28s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/a8e41d46463649ce996f38b431f9681b : SUCCESS in 1h 03m 00s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/e5c4f7eb6be046fa84fde4b875576a63 : SUCCESS in 58m 57s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/7546edcce6ea4a349ada94ffae7b0d48 : SUCCESS in 54m 14s","accounts_in_message":[],"_revision_number":1},{"id":"acac7274ad1a05468cb681e98fc0a5f78f2684d0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 06:48:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2197f15daa55f8f7a770be2810216f3d6c79ec99","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 06:49:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4220b2dd3d284d7ebc5bd068ee586058\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1249b03ecae04cf6a6dba4ee90c692eb : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a9370ef261ea03c02e4e8c40bdfe24e97a8692fb","revisions":{"a9370ef261ea03c02e4e8c40bdfe24e97a8692fb":{"kind":"REWORK","_number":1,"created":"2023-03-28 02:16:49.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/32/878732/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/32/878732/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/32/878732/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/32/878732/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/32/878732/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/32/878732/1"}}},"commit":{"parents":[{"commit":"f6347c1ff7a63fa281f12f50684e2fef2077477f","subject":"feat: update the Korean i18n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f6347c1ff7a63fa281f12f50684e2fef2077477f"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-03-28 02:15:57.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-03-28 02:15:57.000000000","tz":480},"subject":"feat: update zh-cn to zh-hans","message":"feat: update zh-cn to zh-hans\n\nUpdate zh-cn to zh-hans, and update the icon when switch language\n\nChange-Id: I3e6d3e8aebec4584f863282082a11b36d7cee193\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a9370ef261ea03c02e4e8c40bdfe24e97a8692fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a9370ef261ea03c02e4e8c40bdfe24e97a8692fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
