)]}'
{"id":"openstack%2Fskyline-console~878769","triplet_id":"openstack%2Fskyline-console~master~I68691c9743aca6196f1e2258cb2acaf2a82066dc","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2023-04-18 09:36:37.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-04-18 08:03:43.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"34680":{"account":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"last_update":"2023-04-18 09:36:37.000000000","reason":"Change was submitted"},"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2023-04-18 09:36:37.000000000","reason":"Change was submitted"},"34589":{"account":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"last_update":"2023-04-18 09:36:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I68691c9743aca6196f1e2258cb2acaf2a82066dc","subject":"Add Designate (DNS) UI to Skyline-Console","status":"MERGED","created":"2023-03-28 12:21:47.000000000","updated":"2023-04-18 09:37:33.000000000","submitted":"2023-04-18 09:36:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":29,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"878769","meta_rev_id":"edf765303d8201450208236ca399de875357bb44","_number":878769,"virtual_id_number":878769,"owner":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-18 09:36:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2023-04-18 08:03:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2023-04-18 08:03:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"}],"reviewers":{"CC":[{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-28 13:07:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-06 08:47:07.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"CC"},{"updated":"2023-04-06 08:47:07.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"state":"REVIEWER"},{"updated":"2023-04-06 08:47:07.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2023-04-06 08:49:08.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"state":"REVIEWER"},{"updated":"2023-04-06 08:50:05.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"},{"updated":"2023-04-06 09:01:10.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"CC"}],"messages":[{"id":"96e669ef43b718ad0b3fce52e9968407607b43d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"date":"2023-03-28 12:21:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c0333d0458b1e242fe665c3c29b2f41ddad478b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 13:07:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/19c1bf9fae0f4a63972d406f4c9751b3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3869b2b0eb4e46af9db69d571643fc3a : SUCCESS in 5m 03s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/25f33a32b06c4121a9ac00d5870fb63d : FAILURE in 10m 24s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/15262fad41384ddc9050d88ed1acfeac : SUCCESS in 6m 29s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/8acdb55faf23447380217aac1943b836 : SUCCESS in 7m 28s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/af4a296ab44c40a9bedb50974bceca49 : FAILURE in 24m 51s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/bcf3ba7e655c4dba989f964444bfd737 : FAILURE in 31m 18s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/b93158c296724ba68a25d7dba9e9c521 : FAILURE in 31m 40s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/337333aecab54c499938cd1ac624281f : FAILURE in 28m 42s","accounts_in_message":[],"_revision_number":1},{"id":"e6d49914911c8f334318e00e00fbfad06628f2d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"date":"2023-03-28 13:34:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7c7a99132e4cba605168f80b01d173495788ac40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 14:50:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8df70e0cfe1428490a65abb7212bb29\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/920bbe8853d44738ba721337b2974d40 : SUCCESS in 4m 51s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/40e4888241d7412ebc2c467b8452669d : FAILURE in 9m 45s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/0bfb0134cb984975a593c580946583c6 : SUCCESS in 7m 38s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/779afdff2d2446b6abb8e2cad433da60 : SUCCESS in 7m 30s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/4a64fc709fc44e9e9bf5d170499c832d : SUCCESS in 1h 12m 17s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/25333b866ae145cf8d9b61d88d3b4f64 : SUCCESS in 1h 03m 15s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/fee123b9f5124606b64578717ef7d208 : SUCCESS in 55m 04s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/8705847e41164d57b50ac818aaafc795 : SUCCESS in 47m 05s","accounts_in_message":[],"_revision_number":2},{"id":"6dbad42c41ad73beca432e7d1e91e054662f4a6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"date":"2023-03-29 05:43:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8b52b054b6c1064aa1301c3605cc7ffc35b24108","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 07:23:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77c7ac1ebc1d4c8688d79fcddffc71cd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c675df0a14541e2bb22993e6bd70d95 : SUCCESS in 4m 04s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/287ef4bb92654f529793b55417dfaefe : SUCCESS in 10m 22s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/78b531ee00514d2b941808191c5e129f : SUCCESS in 4m 40s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/5b5b00f26523415ab89aad4a17653542 : SUCCESS in 7m 02s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/653160184d5f4b0a9c7ae76452138574 : SUCCESS in 1h 34m 33s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/ca00c0e126504b92b000a3332683d13b : SUCCESS in 50m 58s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/e08f6373d58f41ea80d0f2ff4a268661 : SUCCESS in 57m 50s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/963207b4257e43c694478d4c45cbc37f : SUCCESS in 45m 19s","accounts_in_message":[],"_revision_number":3},{"id":"504e7599b1ffc9cb9b1b70459754da70278c2498","author":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"date":"2023-04-03 05:58:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"16abc2790b9e3ac1db869b589b936e42dd49950e","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-04-06 08:59:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"10e8838fad89c8c118bdd9b81882e52637882c71","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-04-06 10:05:31.000000000","message":"Patch Set 3:\n\n(13 comments)","accounts_in_message":[],"_revision_number":3},{"id":"969ef7d07e2683280be62b8c45712e11ed9913f6","author":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"date":"2023-04-10 08:01:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"46f570b4b59e7fbcb1f86dd60e006587df1a825a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"date":"2023-04-10 13:04:19.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"2cf50dadc44d63e0fa858ac529a31335e14b02de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"date":"2023-04-10 13:14:02.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"db3491f85b0e015eecf02210712e8e0600f0a40e","author":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"date":"2023-04-10 13:15:21.000000000","message":"Patch Set 5:\n\n(12 comments)","accounts_in_message":[],"_revision_number":5},{"id":"9770a888cc8f0456e5f99631977358eb7a545367","author":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"date":"2023-04-10 13:15:40.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"01cf8e4862a50e5b99650926e65a289dfa457319","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-10 14:46:36.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f73415d050047a9a2d978f29b9b0bbe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12ba00e0eeb944d5bff367f39b8d88e3 : SUCCESS in 4m 28s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/8c1a3de092284a248f7a3f72abd635a3 : SUCCESS in 11m 51s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/331a9be63cec4c42b1cc5cd65b20af7a : SUCCESS in 5m 45s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/4b5b361b1ee248ff9552e48ace29e5bd : SUCCESS in 6m 57s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/98044e41ec22429c89c86c2426207fd7 : SUCCESS in 1h 20m 20s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/f3709be85eb94367910af97142e13201 : SUCCESS in 1h 02m 45s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/ae4796c4e1074485b4fe9e89cc9f7283 : SUCCESS in 43m 32s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/017b58d9e89349efb7ebc9f700aa9def : SUCCESS in 56m 42s","accounts_in_message":[],"_revision_number":5},{"id":"40082f8317bf554165eb67c7b63fef90079cffbc","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-04-18 08:03:43.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"f945fe36561ce2d38ddb6df9fc09535a6253fd2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 08:04:21.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"50b47e7b0427e3f302228bb148453a9c1d6e2bd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 09:36:36.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28fabeee22384c0892e3122227a89103\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90e6caf9926e433cadc926cef2f903d6 : SUCCESS in 5m 30s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/d62251f665514bb7bc16f784a0d2c329 : SUCCESS in 8m 25s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/66b4d2427ee048d5b6368977e9bf6b77 : SUCCESS in 5m 28s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/8762142c7ecb4c17b2ff606c9f932723 : SUCCESS in 7m 48s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/a6a6ceb364ac49f3a469eb47afd825df : SUCCESS in 1h 27m 16s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/fef44a8691ad4467a1551b69b1bb05c5 : SUCCESS in 1h 03m 02s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/6630398d3a0d43218536d15a1c528917 : SUCCESS in 53m 04s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/edd3db0365664e0aadf5dcc82f6dadb5 : SUCCESS in 47m 10s","accounts_in_message":[],"_revision_number":5},{"id":"20d9d2e383ae1450012b8ba41a8f978d6ed2ae36","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 09:36:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"edf765303d8201450208236ca399de875357bb44","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 09:37:33.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e537fce28f444f43bd3802180069deff\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/695a9fa91139492482b95bf1cc78e2d7 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"ce20a90c1e29df3285cb710639ab04bbe902859c","revisions":{"0e86e57e1e3c545fc7ab2eb36772c7403ea7284d":{"kind":"REWORK","_number":1,"created":"2023-03-28 12:21:47.000000000","uploader":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"ref":"refs/changes/69/878769/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/69/878769/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/69/878769/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/69/878769/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/69/878769/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/69/878769/1"}}},"commit":{"parents":[{"commit":"a9370ef261ea03c02e4e8c40bdfe24e97a8692fb","subject":"feat: update zh-cn to zh-hans","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a9370ef261ea03c02e4e8c40bdfe24e97a8692fb"}]}],"author":{"name":"Numan MENEKŞE","email":"numan.menekse@tubitak.gov.tr","date":"2023-03-28 12:17:14.000000000","tz":0},"committer":{"name":"Numan MENEKŞE","email":"numan.menekse@tubitak.gov.tr","date":"2023-03-28 12:21:12.000000000","tz":0},"subject":"Add Designate (DNS) UI to Skyline-Console","message":"Add Designate (DNS) UI to Skyline-Console\n\nChange-Id: I68691c9743aca6196f1e2258cb2acaf2a82066dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0e86e57e1e3c545fc7ab2eb36772c7403ea7284d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0e86e57e1e3c545fc7ab2eb36772c7403ea7284d"}]},"branch":"refs/heads/master"},"ca2a346becf2d440ad0b24f29ce28487fb3c070a":{"kind":"REWORK","_number":2,"created":"2023-03-28 13:34:20.000000000","uploader":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"ref":"refs/changes/69/878769/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/69/878769/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/69/878769/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/69/878769/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/69/878769/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/69/878769/2"}}},"commit":{"parents":[{"commit":"a9370ef261ea03c02e4e8c40bdfe24e97a8692fb","subject":"feat: update zh-cn to zh-hans","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a9370ef261ea03c02e4e8c40bdfe24e97a8692fb"}]}],"author":{"name":"Numan MENEKŞE","email":"numan.menekse@tubitak.gov.tr","date":"2023-03-28 12:17:14.000000000","tz":0},"committer":{"name":"Numan MENEKŞE","email":"numan.menekse@tubitak.gov.tr","date":"2023-03-28 13:33:29.000000000","tz":0},"subject":"Add Designate (DNS) UI to Skyline-Console","message":"Add Designate (DNS) UI to Skyline-Console\n\nChange-Id: I68691c9743aca6196f1e2258cb2acaf2a82066dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ca2a346becf2d440ad0b24f29ce28487fb3c070a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ca2a346becf2d440ad0b24f29ce28487fb3c070a"}]},"branch":"refs/heads/master"},"c16dfc2ae44660861870fa2f18fb349d071e7406":{"kind":"REWORK","_number":3,"created":"2023-03-29 05:43:28.000000000","uploader":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"ref":"refs/changes/69/878769/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/69/878769/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/69/878769/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/69/878769/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/69/878769/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/69/878769/3"}}},"commit":{"parents":[{"commit":"a9370ef261ea03c02e4e8c40bdfe24e97a8692fb","subject":"feat: update zh-cn to zh-hans","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a9370ef261ea03c02e4e8c40bdfe24e97a8692fb"}]}],"author":{"name":"Numan MENEKŞE","email":"numan.menekse@tubitak.gov.tr","date":"2023-03-28 12:17:14.000000000","tz":0},"committer":{"name":"Numan MENEKŞE","email":"numan.menekse@tubitak.gov.tr","date":"2023-03-29 05:42:10.000000000","tz":0},"subject":"Add Designate (DNS) UI to Skyline-Console","message":"Add Designate (DNS) UI to Skyline-Console\n\nChange-Id: I68691c9743aca6196f1e2258cb2acaf2a82066dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c16dfc2ae44660861870fa2f18fb349d071e7406"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c16dfc2ae44660861870fa2f18fb349d071e7406"}]},"branch":"refs/heads/master"},"0f6a464b8ee00211ae8abd1f4067e1b309ad7c10":{"kind":"REWORK","_number":4,"created":"2023-04-10 13:04:19.000000000","uploader":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"ref":"refs/changes/69/878769/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/69/878769/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/69/878769/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/69/878769/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/69/878769/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/69/878769/4"}}},"commit":{"parents":[{"commit":"a9370ef261ea03c02e4e8c40bdfe24e97a8692fb","subject":"feat: update zh-cn to zh-hans","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a9370ef261ea03c02e4e8c40bdfe24e97a8692fb"}]}],"author":{"name":"Numan MENEKŞE","email":"numan.menekse@tubitak.gov.tr","date":"2023-03-28 12:17:14.000000000","tz":0},"committer":{"name":"Numan MENEKŞE","email":"numan.menekse@tubitak.gov.tr","date":"2023-04-10 12:27:28.000000000","tz":0},"subject":"Add Designate (DNS) UI to Skyline-Console","message":"Add Designate (DNS) UI to Skyline-Console\n\nChange-Id: I68691c9743aca6196f1e2258cb2acaf2a82066dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0f6a464b8ee00211ae8abd1f4067e1b309ad7c10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0f6a464b8ee00211ae8abd1f4067e1b309ad7c10"}]},"branch":"refs/heads/master"},"ce20a90c1e29df3285cb710639ab04bbe902859c":{"kind":"REWORK","_number":5,"created":"2023-04-10 13:14:02.000000000","uploader":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"ref":"refs/changes/69/878769/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/69/878769/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/69/878769/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/69/878769/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/69/878769/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/69/878769/5"}}},"commit":{"parents":[{"commit":"a9370ef261ea03c02e4e8c40bdfe24e97a8692fb","subject":"feat: update zh-cn to zh-hans","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a9370ef261ea03c02e4e8c40bdfe24e97a8692fb"}]}],"author":{"name":"Numan MENEKŞE","email":"numan.menekse@tubitak.gov.tr","date":"2023-03-28 12:17:14.000000000","tz":0},"committer":{"name":"Numan MENEKŞE","email":"numan.menekse@tubitak.gov.tr","date":"2023-04-10 13:11:40.000000000","tz":0},"subject":"Add Designate (DNS) UI to Skyline-Console","message":"Add Designate (DNS) UI to Skyline-Console\n\nChange-Id: I68691c9743aca6196f1e2258cb2acaf2a82066dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ce20a90c1e29df3285cb710639ab04bbe902859c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ce20a90c1e29df3285cb710639ab04bbe902859c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
