)]}'
{"id":"openstack%2Fskyline-console~880793","triplet_id":"openstack%2Fskyline-console~master~I88f3a842bcf3a91d1cd8a0f20e847f1e5e57ed7d","project":"openstack/skyline-console","branch":"master","topic":"select-table","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-04-19 10:10:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I88f3a842bcf3a91d1cd8a0f20e847f1e5e57ed7d","subject":"feat: optimize the clicking event in the select-table","status":"MERGED","created":"2023-04-19 06:38:05.000000000","updated":"2023-04-19 10:11:37.000000000","submitted":"2023-04-19 10:10:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880793-select-table","meta_rev_id":"3f8775dd7624cd00a52da58257481f6e7d774935","_number":880793,"virtual_id_number":880793,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-19 10:10:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-04-19 08:55:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":1,"date":"2023-04-19 08:53:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-04-19 08:55:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-19 08:48:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-19 08:53:40.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"state":"REVIEWER"},{"updated":"2023-04-19 08:55:04.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"84fed215070add43152c9659c07c8d3f41ccccec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-04-19 06:38:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3fd012e523d9791fa09f9306183058c52d92a219","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-04-19 06:45:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"83f0da0152d09e5c77926c55219a94020d200b1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 08:48:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2868241639064962b8b3815f8832b543\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/140705c5cbf44e59b2112c2f380c19cd : SUCCESS in 5m 43s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/017e0bd92ebd48848ca6fdec324e14dd : SUCCESS in 11m 23s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/d89b88b4aa024297804ebc5490d52371 : SUCCESS in 6m 15s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/5ec8ca89f4914670b7829f5d5452bf86 : SUCCESS in 7m 38s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/7fc52a099b72469d9df4f8fd6ce5c60f : SUCCESS in 1h 26m 55s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/31566d929d484cedb72314771c57e3d7 : SUCCESS in 1h 47m 29s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/24bcad28a4a9448dbec4d111e7150a7f : SUCCESS in 53m 04s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/a465c45d81e34d7a9092646f9a3cac66 : SUCCESS in 58m 19s","accounts_in_message":[],"_revision_number":2},{"id":"6cd3a8158c87b3cbceb5c999699021eabe09c0a1","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-04-19 08:53:40.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"6df8a97366aa3f0fdf725a75d8d40a5ef2e952fb","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-04-19 08:55:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6696e934b29500996ac67da001bbb4b6a9fc4835","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 08:55:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f680d4730acea12598eceaacb087f5aa82ddf3a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 10:10:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7e3cc445895464e8c4e6859cc68e804\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca444609fde7494b95f092dcdea6ad96 : SUCCESS in 5m 10s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/026b78eb1bee458f9ae0647c0128629f : SUCCESS in 10m 07s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/58dc7d3f1e1440a988c3ee6ae1788b15 : SUCCESS in 6m 48s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/4988e0eda16044c4b313fbe08ef44a91 : SUCCESS in 7m 37s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/08e7cf957b134e52a5dd484af8c786ca : SUCCESS in 1h 10m 21s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/fde58d4edb36441d9fd7194eb27aaed5 : SUCCESS in 1h 07m 53s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/66c12114b9dd4ea687cb0e41d778a204 : SUCCESS in 1h 02m 15s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/9154586d7fd2449aa675d59c6e74d548 : SUCCESS in 51m 15s","accounts_in_message":[],"_revision_number":2},{"id":"c348b11c747f9d40bbf9e4ac1aabf1a4f86b601a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 10:10:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"3f8775dd7624cd00a52da58257481f6e7d774935","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 10:11:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aec61b0bcb784415bd26fead99ef3bbf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7b1155282b545e5971d94e8c6e59cee : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"56230e524e48b77ccd7672f0c18a0c71499d2be5","revisions":{"ee2bb0ffba0b552b8b1feeb94c5b1a9c06d642f1":{"kind":"REWORK","_number":1,"created":"2023-04-19 06:38:05.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/93/880793/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/93/880793/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/880793/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/880793/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/880793/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/93/880793/1"}}},"commit":{"parents":[{"commit":"c73fd49f225745a68f428cb7f816ff35504b8f06","subject":"Merge \"Add Designate (DNS) UI to Skyline-Console\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c73fd49f225745a68f428cb7f816ff35504b8f06"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-04-19 06:37:21.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-04-19 06:37:42.000000000","tz":480},"subject":"feat: optimize the clicking event in the select-table","message":"feat: optimize the clicking event in the select-table\n\nOptimize the clicking event in the select-table component, when click the row of the table in the component, the row will be selected, but when click the button in the row, the selection does not change.\n\nChange-Id: I88f3a842bcf3a91d1cd8a0f20e847f1e5e57ed7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ee2bb0ffba0b552b8b1feeb94c5b1a9c06d642f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ee2bb0ffba0b552b8b1feeb94c5b1a9c06d642f1"}]},"branch":"refs/heads/master"},"56230e524e48b77ccd7672f0c18a0c71499d2be5":{"kind":"REWORK","_number":2,"created":"2023-04-19 06:45:27.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/93/880793/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/93/880793/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/880793/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/880793/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/880793/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/93/880793/2"}}},"commit":{"parents":[{"commit":"c73fd49f225745a68f428cb7f816ff35504b8f06","subject":"Merge \"Add Designate (DNS) UI to Skyline-Console\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c73fd49f225745a68f428cb7f816ff35504b8f06"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-04-19 06:37:21.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-04-19 06:45:08.000000000","tz":480},"subject":"feat: optimize the clicking event in the select-table","message":"feat: optimize the clicking event in the select-table\n\nOptimize the clicking event in the select-table component, when click the row of the table in the component, the row will be selected, but when click the button in the row, the selection does not change.\n\nChange-Id: I88f3a842bcf3a91d1cd8a0f20e847f1e5e57ed7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/56230e524e48b77ccd7672f0c18a0c71499d2be5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/56230e524e48b77ccd7672f0c18a0c71499d2be5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
