)]}'
{"id":"openstack%2Fskyline-console~880807","triplet_id":"openstack%2Fskyline-console~master~If639e04f5c3327eda1bdce3652f013f628e0c1b5","project":"openstack/skyline-console","branch":"master","topic":"select-table","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-04-21 03:52:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If639e04f5c3327eda1bdce3652f013f628e0c1b5","subject":"feat: support clear all selected data for select-table","status":"MERGED","created":"2023-04-19 10:21:46.000000000","updated":"2023-04-21 03:53:58.000000000","submitted":"2023-04-21 03:52:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880807-select-table","meta_rev_id":"d4116b0788cb4e095b5844cd35c7fa50f4b7a3b0","_number":880807,"virtual_id_number":880807,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-21 03:52:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-04-21 02:15:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":1,"date":"2023-04-21 01:57:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-04-21 02:15:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-19 11:44:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-21 01:57:37.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"state":"REVIEWER"},{"updated":"2023-04-21 02:15:46.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"c36bd454ec9ad9f818ff7a84943b0178062dd8b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-04-19 10:21:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"166087fd912009775795df8bb5e887bc6f16727c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-04-19 10:30:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"63396c94f422695573bc06e43787c1d41d4ba361","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 11:44:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9533d757d1044bd3a63ecfdbf9994775\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21d5b0e4bace4f70a11998a76f9aef4a : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0331b83afdd04b1394560e2d0b3306a9 : SUCCESS in 5m 44s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/f2d427fcc4bf4fec88c42651e73fb38d : SUCCESS in 9m 41s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/af049b5234a64fd492505a36c350d61c : SUCCESS in 5m 16s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/202e67dfb70d4f01b03e83844195c8f7 : SUCCESS in 7m 27s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/0ca6100ebed84200b9ac0e14bf04216b : SUCCESS in 1h 10m 53s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/370470e4afcc40f6bc3f28dc73a6873e : SUCCESS in 1h 10m 54s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/84a99f199c6f4deaa31af7ccc7e6db2e : SUCCESS in 47m 13s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/685741cc827147ccb66b0efd92cc851b : SUCCESS in 38m 48s","accounts_in_message":[],"_revision_number":2},{"id":"acfbb28b472b18702058d48c4c2fa537d50e0e91","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-04-21 01:57:37.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9c22c927371b10182c23e83f26a7180bbcd86654","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-04-21 02:15:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e5ea2ab83fa7a7c56d07c517b88757eb4878bc9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 02:16:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"127a474b6cd0b28836f462c05b4727e39cda7c53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 03:52:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8e1f39eb56e4ee5b6b8865a0e71d084\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/924846c576f940edba014f5b0d9f4cf4 : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/950512a66ffd4ad2a5fe31ca20be85c4 : SUCCESS in 4m 46s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/918a03b6231041009b556d066854ecd3 : SUCCESS in 11m 24s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/6f201bf81659431fa5153a17b76bc491 : SUCCESS in 7m 37s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/786560df29554976813952d72488e235 : SUCCESS in 7m 09s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/4fef015598834ff993ff10a87288c71f : SUCCESS in 1h 29m 32s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/0712f99cfff342d3b73f8a5cfe20b3a6 : SUCCESS in 57m 21s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/aa52ece177d34184bc3089cd67c25e4c : SUCCESS in 56m 36s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/00e25ff29d894afdac152f031cec8bf7 : SUCCESS in 50m 48s","accounts_in_message":[],"_revision_number":2},{"id":"0baca0b3f3be8a941cf2679fc4b97cccce20ac62","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 03:52:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d4116b0788cb4e095b5844cd35c7fa50f4b7a3b0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 03:53:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/326d7223b4854a9997b024cc4252e781\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3be0f375fdab4143a205dcab2800460a : SUCCESS in 41s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45cd9dce360d409eae3472eca2084e6e : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b30e7ae9a35c20edb0864669b50b43e698830081","revisions":{"2d987ae3f77e2084cd41c4cee4c8135df09f728e":{"kind":"REWORK","_number":1,"created":"2023-04-19 10:21:46.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/07/880807/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/880807/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/880807/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/880807/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/880807/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/880807/1"}}},"commit":{"parents":[{"commit":"56230e524e48b77ccd7672f0c18a0c71499d2be5","subject":"feat: optimize the clicking event in the select-table","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/56230e524e48b77ccd7672f0c18a0c71499d2be5"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-04-19 08:00:27.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-04-19 10:21:31.000000000","tz":480},"subject":"feat: support clear all selected data for select-table","message":"feat: support clear all selected data for select-table\n\nSupport clear all selected data for the select-table and tab-select-table component.\n\nChange-Id: If639e04f5c3327eda1bdce3652f013f628e0c1b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2d987ae3f77e2084cd41c4cee4c8135df09f728e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2d987ae3f77e2084cd41c4cee4c8135df09f728e"}]},"branch":"refs/heads/master"},"b30e7ae9a35c20edb0864669b50b43e698830081":{"kind":"REWORK","_number":2,"created":"2023-04-19 10:30:44.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/07/880807/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/880807/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/880807/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/880807/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/880807/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/880807/2"}}},"commit":{"parents":[{"commit":"56230e524e48b77ccd7672f0c18a0c71499d2be5","subject":"feat: optimize the clicking event in the select-table","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/56230e524e48b77ccd7672f0c18a0c71499d2be5"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-04-19 08:00:27.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-04-19 10:30:18.000000000","tz":480},"subject":"feat: support clear all selected data for select-table","message":"feat: support clear all selected data for select-table\n\nSupport clear all selected data for the select-table and tab-select-table component.\n\nChange-Id: If639e04f5c3327eda1bdce3652f013f628e0c1b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b30e7ae9a35c20edb0864669b50b43e698830081"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b30e7ae9a35c20edb0864669b50b43e698830081"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
