)]}'
{"id":"openstack%2Fskyline-console~881691","triplet_id":"openstack%2Fskyline-console~master~Idf906bc8a20d5388fc2da37904ce84c6bf9cfc82","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-04-27 06:00:44.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2023-04-27 07:14:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idf906bc8a20d5388fc2da37904ce84c6bf9cfc82","subject":"fix: Do not remove if the listener has an SNI certificate installed","status":"MERGED","created":"2023-04-27 04:38:06.000000000","updated":"2023-04-27 07:15:35.000000000","submitted":"2023-04-27 07:14:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"881691","meta_rev_id":"3819b517c2273696baa703345c45e05ccf5a8420","_number":881691,"virtual_id_number":881691,"owner":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-27 07:14:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2023-04-27 06:00:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2023-04-27 06:00:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-27 05:46:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-27 05:56:59.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"f4d3ff7bf11b409de7d026547af3b96cdda8acc8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-04-27 04:38:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d26e791bb0510b215d72e894d313f29a8bfd631","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 05:46:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfb2495e03fb491f8047b4310f476794\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cf3387fcb4648a7a109da2af74e2a65 : SUCCESS in 4m 54s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/08e43f6cf5214f358ed644f3e8449a9a : SUCCESS in 10m 22s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/915f1509c6074077ba1828e7af934c7c : SUCCESS in 7m 06s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/8f6ea26a5506419b8677d138c7fd3643 : SUCCESS in 7m 08s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/de444b0f8bf9466da2d94ad9a777572a : SUCCESS in 1h 07m 51s","accounts_in_message":[],"_revision_number":1},{"id":"c4fe65fb0c14272f47f29b391971a4083311c88b","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-04-27 06:00:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"130965444b14bfe507668b58a9b6eb65f52f5068","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 06:01:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7699944d2c3783013d8e9c453d7dcae1984422b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 07:14:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/262c2649b818468682f5e1d80b561fec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d9d5d6cc292476596c07457293622b0 : SUCCESS in 5m 16s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/40b6fbee0ea04aa989f13b0de8e2b5fe : SUCCESS in 10m 13s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/72f4e1ab8f7740a2807cb02d8eca5ab5 : SUCCESS in 7m 18s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/e14d27d2d8cc4fafb679fd7277b3b3ed : SUCCESS in 7m 14s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/d6f1008ad6284e1bbfe2fe662f651569 : SUCCESS in 1h 06m 42s","accounts_in_message":[],"_revision_number":1},{"id":"a61ce91bd7032b9f480dd9fa49dfdce0abf5861c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 07:14:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3819b517c2273696baa703345c45e05ccf5a8420","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 07:15:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1170f6fab7ad4c2a850cddc5796e8fba\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2496865ba294c45bf544c9a4b523313 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1f8c96a5dce3da201f66c070766eb1c50cc827c1","revisions":{"1f8c96a5dce3da201f66c070766eb1c50cc827c1":{"kind":"REWORK","_number":1,"created":"2023-04-27 04:38:06.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/91/881691/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/91/881691/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/91/881691/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/91/881691/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/91/881691/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/91/881691/1"}}},"commit":{"parents":[{"commit":"4817423f801440cc2ad224566449be2f0ded3dd5","subject":"Merge \"fix: fix close the error message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4817423f801440cc2ad224566449be2f0ded3dd5"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-04-27 04:35:26.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-04-27 04:35:38.000000000","tz":480},"subject":"fix: Do not remove if the listener has an SNI certificate installed","message":"fix: Do not remove if the listener has an SNI certificate installed\n\n1. Disable sni enabled when edit if sni existed\n2. Update translate for i18n\n\nCloses-Bug: #2017844\nChange-Id: Idf906bc8a20d5388fc2da37904ce84c6bf9cfc82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1f8c96a5dce3da201f66c070766eb1c50cc827c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1f8c96a5dce3da201f66c070766eb1c50cc827c1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
