)]}'
{"id":"openstack%2Fskyline-console~882235","triplet_id":"openstack%2Fskyline-console~master~Ie0cf7b8121065f27f8f16ad48dd290a0eca02fd7","project":"openstack/skyline-console","branch":"master","topic":"bugs","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-05-04 09:41:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie0cf7b8121065f27f8f16ad48dd290a0eca02fd7","subject":"fix: better display cluster health status","status":"MERGED","created":"2023-05-04 06:58:59.000000000","updated":"2023-05-04 09:42:27.000000000","submitted":"2023-05-04 09:41:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"882235-bugs","meta_rev_id":"bf2f4834845ccc9242e39558127bd8bb16b01272","_number":882235,"virtual_id_number":882235,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-04 09:41:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-05-04 08:48:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":1,"date":"2023-05-04 08:43:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-05-04 08:48:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-04 07:42:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-04 08:43:15.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"state":"REVIEWER"},{"updated":"2023-05-04 08:48:29.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"ddb744ac3ca5d48b491b145fc18209520d1cfeea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-05-04 06:58:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f92a4d59f56afa6e7843e85b95bf33aba444f05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 07:42:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1aa6a73abc714abf97dc8a6a7b682df5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bfe34ea8af34aceb44fa419fe4ca34c : SUCCESS in 4m 58s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/ae34e884460b4148a5125cb5f30b556d : SUCCESS in 10m 56s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/9aa6dc5f094c40678e634d0e48aaac41 : SUCCESS in 7m 27s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/3293dc65457a4afd90c5c4f8b9f4c37f : SUCCESS in 7m 51s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/78c938eead924a8dbb29e9104f42f402 : SUCCESS in 42m 02s","accounts_in_message":[],"_revision_number":1},{"id":"97141b858521c59f516317935ad1c0c2e406bfe4","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-05-04 08:43:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"65fafed7a21dc7f2cec91641416fc81668974f4e","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-05-04 08:48:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5c18275b4eb3c9b36d34f2cf96debec8fbf7c162","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 08:49:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"931b864111acb6d296539cfeb9edee05b1db117e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 09:41:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49c773ce8c284d6681146778af81529f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c0011e50e374d4e8707b079e4b61b8f : SUCCESS in 5m 46s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/6caf47338cd04f5698e2d0a7b62a6529 : SUCCESS in 9m 42s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/4099f39ba0b441189d6484e873ef3968 : SUCCESS in 8m 43s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/e4eb5291a7434de0b63c30d313dba47b : SUCCESS in 5m 34s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/c8f9a3381e66435788cc399d7b7ec645 : SUCCESS in 51m 40s","accounts_in_message":[],"_revision_number":1},{"id":"02349b653e8d4023aeb55f425dc40e06f06a8fdd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 09:41:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bf2f4834845ccc9242e39558127bd8bb16b01272","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 09:42:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3939c8a404244a5bfb12574f67f1afd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e924c46b2b394549a82510e3bdc34117 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5c85078d8d6878e49d4365a9a857386532732f8d","revisions":{"5c85078d8d6878e49d4365a9a857386532732f8d":{"kind":"REWORK","_number":1,"created":"2023-05-04 06:58:59.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/35/882235/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/35/882235/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/35/882235/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/35/882235/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/35/882235/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/35/882235/1"}}},"commit":{"parents":[{"commit":"2939a504f1cefc3834be6a4da6c5791f858dbb8a","subject":"fix: fix the port type display","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2939a504f1cefc3834be6a4da6c5791f858dbb8a"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-05-04 06:58:40.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-05-04 06:58:40.000000000","tz":480},"subject":"fix: better display cluster health status","message":"fix: better display cluster health status\n\nUse text to display the health status of the cluster instead of using the Status component\n\nChange-Id: Ie0cf7b8121065f27f8f16ad48dd290a0eca02fd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5c85078d8d6878e49d4365a9a857386532732f8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5c85078d8d6878e49d4365a9a857386532732f8d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
