)]}'
{"id":"openstack%2Fskyline-console~882369","triplet_id":"openstack%2Fskyline-console~master~I932ce902f146072b46c05519e3e1f94db6556ff0","project":"openstack/skyline-console","branch":"master","topic":"bugs","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-05-05 10:13:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I932ce902f146072b46c05519e3e1f94db6556ff0","subject":"fix: optimize the Transfer","status":"MERGED","created":"2023-05-05 02:47:45.000000000","updated":"2023-05-05 10:14:40.000000000","submitted":"2023-05-05 10:13:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"882369-bugs","meta_rev_id":"9005cbc2c5b75152a6fe32c4c876d0fe31dddbe8","_number":882369,"virtual_id_number":882369,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-05 10:13:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-05-05 09:05:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":1,"date":"2023-05-05 08:23:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-05-05 09:05:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-05 03:25:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-05 08:23:56.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"state":"REVIEWER"},{"updated":"2023-05-05 09:05:51.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"515549ade814d80c385b8f255bae93dfe77ff616","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-05-05 02:47:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4582a9eb8ae48ca87163975ecabcc0ee8fb98b4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 03:25:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/056cde7b15824002bb208bf42f9ecc52\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8abae335bfad47799766779ed7601657 : SUCCESS in 4m 36s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/470bd211c56448edbd543564e0c2b819 : SUCCESS in 9m 07s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/26f44487760f40c7a6cb796f38872006 : SUCCESS in 7m 34s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/77ffa0f26af443e280e17a83e37d4e8d : SUCCESS in 6m 18s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/735b84b3e64b4cc497f129d8371b10d5 : FAILURE in 29m 42s","accounts_in_message":[],"_revision_number":1},{"id":"90dbd0b31d04d76e344a69f954bbd4e91293c0aa","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-05-05 06:08:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c8c9994b9f915329c4f57e02ca1d7b98d2388625","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 07:09:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf20f96f487c44afaabde7d0cd04d5ca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/719a5635a53b473ea5130b3cb7368f74 : SUCCESS in 4m 55s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/52f9b838f39a45ff8eaafd1e365eabb9 : SUCCESS in 7m 46s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/fa07c9e445984541ac351f4af6fabc00 : SUCCESS in 7m 21s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/4758b435d6534daa89dddf05897a17ee : SUCCESS in 7m 20s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/090ae3fe845446bcaffcb8ea999d098a : SUCCESS in 50m 31s","accounts_in_message":[],"_revision_number":1},{"id":"fc24046dcc6270debaa1e7158fb974f92b649d2a","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-05-05 08:23:56.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e1133b9c802d729a180bfdc2dc25f35f46ef1970","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-05-05 09:05:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f2bb1d145475574aff4ac80b450654c1adec42d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 09:06:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ad9f650800ae6635ed1054c68ecfa572e033c128","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 10:13:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ffc735286794b76ac136842dfb747e6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e356d13d8b8b4736913fd239a16c6eaf : SUCCESS in 9m 46s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/faf4ea406f2242dba3feb3baea3532a2 : SUCCESS in 8m 35s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/17dac04ea8f64bb39a44efa08a53e24a : SUCCESS in 8m 25s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/5ca5adb65f3243229917a6ec5d2b7512 : SUCCESS in 5m 56s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/88f4509003ea4b88b9cb12d5b40f2ac8 : SUCCESS in 1h 06m 03s","accounts_in_message":[],"_revision_number":1},{"id":"73d0064f865f226b0d3b55c1b6ce9c8676ac0e70","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 10:13:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9005cbc2c5b75152a6fe32c4c876d0fe31dddbe8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 10:14:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17b93f701c5f452dbe8acd48aea3440f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c5fc9f42314460aa6e83c2d15463cbb : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"614d837075f6a489966b3baaa3ff46d8961de004","revisions":{"614d837075f6a489966b3baaa3ff46d8961de004":{"kind":"REWORK","_number":1,"created":"2023-05-05 02:47:45.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/69/882369/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/69/882369/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/69/882369/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/69/882369/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/69/882369/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/69/882369/1"}}},"commit":{"parents":[{"commit":"ce254bd808ea179c3fc44635e5ded072dff2972c","subject":"fix: fix the menu title","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ce254bd808ea179c3fc44635e5ded072dff2972c"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-05-05 02:43:43.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-05-05 02:43:43.000000000","tz":480},"subject":"fix: optimize the Transfer","message":"fix: optimize the Transfer\n\nOptimize the UE of the Transfer component, when click the row of the right table in the Transfer, the row\u0027s selection can be changed. Optimized operations: project manage user, project manage user group, set project when create user\n\nChange-Id: I932ce902f146072b46c05519e3e1f94db6556ff0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/614d837075f6a489966b3baaa3ff46d8961de004"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/614d837075f6a489966b3baaa3ff46d8961de004"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
