)]}'
{"id":"openstack%2Fskyline-console~882636","triplet_id":"openstack%2Fskyline-console~master~I78caf5d172ee51b6f4f6194901f3de4d759c960e","project":"openstack/skyline-console","branch":"master","topic":"bugs","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-05-09 07:53:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I78caf5d172ee51b6f4f6194901f3de4d759c960e","subject":"feat: set default os_type","status":"MERGED","created":"2023-05-09 02:46:12.000000000","updated":"2023-05-09 07:54:26.000000000","submitted":"2023-05-09 07:53:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"882636-bugs","meta_rev_id":"ea0bc4169503d952220774b236e31f3a6233a506","_number":882636,"virtual_id_number":882636,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-09 07:53:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-05-09 06:18:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":1,"date":"2023-05-09 06:14:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-05-09 06:18:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-09 04:54:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-09 06:14:23.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"state":"REVIEWER"},{"updated":"2023-05-09 06:18:38.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"1c7bdb455df6f862340ac4980dc7ed88d1f06df1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-05-09 02:46:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc0f92cb24837841f56a0bbc01d1aa627ab4a84a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 04:54:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b98378905db14a8fa48a52b6f4e2ff85\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5b7ceedbde74f44b3e71b1f9b3174e7 : SUCCESS in 5m 11s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/f6b88c50523f4433b0099b55bbd22f78 : SUCCESS in 9m 03s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/11029078663046f7bb79837856e68a66 : SUCCESS in 6m 49s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/aa7e46881f814961b35dfab0391e858a : SUCCESS in 7m 37s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/764bb0f2dbe44f1e8bf46e96fe788e8d : SUCCESS in 2h 00m 01s","accounts_in_message":[],"_revision_number":1},{"id":"acf7aef4996aa92441f1f2105fe8449d5321b6fc","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-05-09 06:14:23.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"10921d80aa38d41083ec3c71dbeae97e4693be8f","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-05-09 06:18:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e1990717d170c443637bd5dfdc3980c4ea3e595c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 06:19:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"def6286883d8e24d7140941ff68462798db529c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 07:53:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e98c18d2d5d4570b1100faf6f062adf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4641526b3d6449eea2b015be7a62b786 : SUCCESS in 4m 18s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/e10fc831e8ef45cd8030eee5d8245615 : SUCCESS in 6m 39s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/037d3f6a744f4a08b99255131a638238 : SUCCESS in 4m 42s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/05c123f9fd6b4b88af78253a82f56d60 : SUCCESS in 4m 51s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/8ca6abacd21e48818e7099b4d45163e2 : SUCCESS in 1h 29m 22s","accounts_in_message":[],"_revision_number":1},{"id":"1355bb37a92dd6178dc707d066fd7edc64802b58","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 07:53:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ea0bc4169503d952220774b236e31f3a6233a506","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 07:54:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7eaf5dd91e8b46ac827a5cc31c0f1cb8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/958c8855313a4740b56c5433161b5dac : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3c5aeaa945627871ea19b8ff4c4ad134c95a2dd0","revisions":{"3c5aeaa945627871ea19b8ff4c4ad134c95a2dd0":{"kind":"REWORK","_number":1,"created":"2023-05-09 02:46:12.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/36/882636/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/36/882636/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/36/882636/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/36/882636/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/36/882636/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/36/882636/1"}}},"commit":{"parents":[{"commit":"482e4ba321842534b2d4fbde3c920957d653a19e","subject":"fix: update the time display","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/482e4ba321842534b2d4fbde3c920957d653a19e"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-05-09 02:41:56.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-05-09 02:41:56.000000000","tz":480},"subject":"feat: set default os_type","message":"feat: set default os_type\n\nSet the image\u0027s os_type\u003dwindows when the image\u0027s os_distro\u003dwindows\n\nChange-Id: I78caf5d172ee51b6f4f6194901f3de4d759c960e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3c5aeaa945627871ea19b8ff4c4ad134c95a2dd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3c5aeaa945627871ea19b8ff4c4ad134c95a2dd0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
