)]}'
{"id":"openstack%2Fskyline-console~882791","triplet_id":"openstack%2Fskyline-console~master~Id985ee76db395a742152d5982af0766559e5a5da","project":"openstack/skyline-console","branch":"master","topic":"bug/2019065","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-05-10 08:01:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id985ee76db395a742152d5982af0766559e5a5da","subject":"fix: fix MagicInput component","status":"MERGED","created":"2023-05-10 02:35:50.000000000","updated":"2023-05-10 08:02:34.000000000","submitted":"2023-05-10 08:01:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"882791-bug/2019065","meta_rev_id":"eb263c2d08d8a76b659ea1e65c7dee5c64ccca6e","_number":882791,"virtual_id_number":882791,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-10 08:01:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-05-10 06:44:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-05-10 06:44:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-10 04:17:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-10 06:44:30.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"23dce6a286703e216093e1154240a95ed977bd66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-05-10 02:35:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b43fe97429bf8e8a922d3607ddfb445b5d45d6fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 04:17:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0786d4825f534266a3bd9ddd5dc6519b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df2969c8b0ad471db16ff6a3601cd832 : SUCCESS in 4m 48s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/99358914bb184a40995a819b2ff2c2a2 : SUCCESS in 8m 20s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/effc8428414e4511bacda2fefddfb252 : SUCCESS in 6m 32s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/b05e1bce67824351a2354657acd1fe5a : SUCCESS in 7m 56s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/c6dda3f5063f4a6a8ee8097b301b91af : SUCCESS in 1h 30m 11s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/17957bae83c64d74b0bbc2109d4a491a : SUCCESS in 1h 05m 25s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/60ad90f8707e4862ab8a49a2474b399a : SUCCESS in 1h 00m 33s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/22850ca5164d44ff8b9c87e8a1631dc5 : FAILURE in 1h 30m 50s","accounts_in_message":[],"_revision_number":1},{"id":"169edbc351b20273b946d5924a8c406b09915e38","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-05-10 05:05:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7ae6f623a7913677028d482b16eb71c17ee19d05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 06:40:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7df87c5457644567b914226e059b55d0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8351074119564dafa323e84a04f43e1a : SUCCESS in 5m 05s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/3bdb094cd0414ecb850bfe6d4305674f : SUCCESS in 10m 14s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/cf239d058abd4701ae4bcad1a6220322 : SUCCESS in 7m 19s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/2e2abeea761e4f65a201bddac2e8931a : SUCCESS in 7m 02s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/bd08da5dc79240f2bdb73865ef9233ae : SUCCESS in 1h 29m 10s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/a186e78c433f456f8c3c160354de5a1c : SUCCESS in 1h 01m 45s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/b5ab1c11ba254a9594cc9f6ab0a92b81 : SUCCESS in 59m 33s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/27a65ca5095249449f6e417bd6a01082 : SUCCESS in 56m 38s","accounts_in_message":[],"_revision_number":1},{"id":"f366ad5403b2057e890374584f1cf6e068a8d68d","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-05-10 06:44:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e53dcb42ec8fb260db6b53f4da99d9fcabec0000","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 06:45:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"229eb5ad985471de161616bc7092bf3c2b4006e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 08:01:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b985d9bdf9e4617a9482e8ff483eae6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69c376be11924cd29c8827ed73ce6220 : SUCCESS in 4m 04s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/7cddd46ecb0a41d696f9db0615956361 : SUCCESS in 10m 22s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/62d4e58440974299bc5c442d5be5fb22 : SUCCESS in 6m 13s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/9c63716629924c8ca5b2af1341cc6d61 : SUCCESS in 5m 22s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/059a32354f604f56b1e698837da14f43 : SUCCESS in 1h 15m 03s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/16d65ddd3dae45b88bb6fdf3ba0d0a3d : SUCCESS in 44m 46s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/aae5ec47f8ac4fb6a96552955f67a4e0 : SUCCESS in 38m 35s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/8fcd403687724c7b902e1be951dcb0de : SUCCESS in 44m 17s","accounts_in_message":[],"_revision_number":1},{"id":"6b5931933a18378a35e554787984bf4a19054f91","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 08:01:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"eb263c2d08d8a76b659ea1e65c7dee5c64ccca6e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 08:02:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3b08862d75344c6a061134757ad1623\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1aabfd8dc4534e65ad6757bfa98924a4 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f395e0e2ec023884af67233dc13c33e7fd2d0c41","revisions":{"f395e0e2ec023884af67233dc13c33e7fd2d0c41":{"kind":"REWORK","_number":1,"created":"2023-05-10 02:35:50.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/91/882791/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/91/882791/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/91/882791/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/91/882791/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/91/882791/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/91/882791/1"}}},"commit":{"parents":[{"commit":"3c5aeaa945627871ea19b8ff4c4ad134c95a2dd0","subject":"feat: set default os_type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3c5aeaa945627871ea19b8ff4c4ad134c95a2dd0"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-05-10 02:17:55.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-05-10 02:33:54.000000000","tz":480},"subject":"fix: fix MagicInput component","message":"fix: fix MagicInput component\n\nFix when the current input value does not generate a tag, but click the clear button, the search menu can not be shown. Clear the input value in the state.\n\nCloses-Bug: #2019065\nChange-Id: Id985ee76db395a742152d5982af0766559e5a5da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f395e0e2ec023884af67233dc13c33e7fd2d0c41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f395e0e2ec023884af67233dc13c33e7fd2d0c41"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
