)]}'
{"id":"openstack%2Fskyline-console~883364","triplet_id":"openstack%2Fskyline-console~master~I421469265e6f12f1583456bc20f8cafa7c48536e","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"35057":{"account":{"_account_id":35057,"name":"JaeSeong-Shin","email":"rornfl916@gmail.com","username":"JaeSeong"},"last_update":"2023-06-27 02:11:28.000000000","reason":"\u003cGERRIT_ACCOUNT_33859\u003e replied on the change","reason_account":{"_account_id":33859,"name":"Yoon Soo Lim","display_name":"yoonsoolim","email":"msdbtjd123@naver.com","username":"yoonsoolim"}},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-07-12 01:58:20.000000000","reason":"Change was submitted"},"14482":{"account":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"last_update":"2023-07-12 01:28:58.000000000","reason":"\u003cGERRIT_ACCOUNT_14482\u003e replied on the change","reason_account":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2023-05-29 05:27:18.000000000","reason":"\u003cGERRIT_ACCOUNT_28706\u003e replied on the change","reason_account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},"33859":{"account":{"_account_id":33859,"name":"Yoon Soo Lim","display_name":"yoonsoolim","email":"msdbtjd123@naver.com","username":"yoonsoolim"},"last_update":"2023-07-12 01:58:20.000000000","reason":"Change was submitted"},"31238":{"account":{"_account_id":31238,"name":"Youngju Lee","email":"leeyj7141@gmail.com","username":"leeyj7141"},"last_update":"2023-05-24 12:04:46.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"},"6282":{"account":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"last_update":"2023-07-12 01:58:20.000000000","reason":"Change was submitted"},"35517":{"account":{"_account_id":35517,"name":"WonGyeongLee","email":"wongyeong.tech@gmail.com","username":"wongyeong"},"last_update":"2023-07-12 01:58:20.000000000","reason":"Change was submitted"},"23279":{"account":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"last_update":"2023-07-12 01:58:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I421469265e6f12f1583456bc20f8cafa7c48536e","subject":"feat: Update i18n Korean translation","status":"MERGED","created":"2023-05-17 12:07:35.000000000","updated":"2023-07-12 01:59:17.000000000","submitted":"2023-07-12 01:58:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883364","meta_rev_id":"7f1de89dcfb1d1b13aa9dd81dc71afa7b8cad10a","_number":883364,"virtual_id_number":883364,"owner":{"_account_id":35517,"name":"WonGyeongLee","email":"wongyeong.tech@gmail.com","username":"wongyeong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-12 01:58:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"recommended":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2023-07-12 01:28:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":2,"date":"2023-06-27 06:26:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":1,"date":"2023-07-12 01:50:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"}],"reviewers":{"CC":[{"_account_id":33859,"name":"Yoon Soo Lim","display_name":"yoonsoolim","email":"msdbtjd123@naver.com","username":"yoonsoolim"}],"REVIEWER":[{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-17 12:16:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-24 12:08:05.000000000","updated_by":{"_account_id":33859,"name":"Yoon Soo Lim","display_name":"yoonsoolim","email":"msdbtjd123@naver.com","username":"yoonsoolim"},"reviewer":{"_account_id":33859,"name":"Yoon Soo Lim","display_name":"yoonsoolim","email":"msdbtjd123@naver.com","username":"yoonsoolim"},"state":"CC"},{"updated":"2023-05-25 20:08:18.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2023-05-25 20:08:58.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"},{"updated":"2023-05-27 10:00:53.000000000","updated_by":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"reviewer":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"state":"REVIEWER"},{"updated":"2023-05-29 05:27:18.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2023-07-12 01:30:00.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"state":"REVIEWER"}],"messages":[{"id":"b3ab7975b14141f96d0c0e1d4924f2adac835ac2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35517,"name":"WonGyeongLee","email":"wongyeong.tech@gmail.com","username":"wongyeong"},"date":"2023-05-17 12:07:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"885555c6f89c99bd14578078cc43866b3db9a210","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 12:16:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83c8f80326264482821782bdc2fc7b5b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d8c94342bc442e182e217b8a57c3ed5 : SUCCESS in 5m 13s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/c74c8b0325014937bf98e0f100470f37 : SUCCESS in 7m 17s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/4a79472d389f4c698d5c937d514c169d : SUCCESS in 6m 25s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/55d73e5076e64bbfa88ecd10dfa19f5e : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":1},{"id":"9519cf7c6c2ea0d016a68dc9e15f3b623e6e3b39","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33859,"name":"Yoon Soo Lim","display_name":"yoonsoolim","email":"msdbtjd123@naver.com","username":"yoonsoolim"},"date":"2023-05-17 12:24:18.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b08bc0c7b649aafbd1e7d890625930c23623892f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 12:41:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d0d65ca486e44d08d1e6569e1201387\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/889da6cdd2b14cfebb821d31a6dc12a9 : SUCCESS in 4m 24s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/6bdd7e3c77114ac0801bb831a9e9edcd : SUCCESS in 10m 03s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/59e24307e6a64e3fad076388584566d4 : SUCCESS in 5m 24s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/58fb9cc5b29249e8a9ddcdc77e540012 : SUCCESS in 6m 42s","accounts_in_message":[],"_revision_number":2},{"id":"44d4b5724a24adb2b69e99a5878e2f58a4a34d8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31238,"name":"Youngju Lee","email":"leeyj7141@gmail.com","username":"leeyj7141"},"date":"2023-05-24 02:21:35.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"4b2962023050c18675c7388658e93520dafdc8ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 02:38:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35690c83ae6d44dbb0fc771bede3c576\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1c28f718cd64d86bfa03b81a016324a : SUCCESS in 4m 20s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/3fc487819b3d4844ac55ecae9b04f52b : SUCCESS in 11m 06s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/e41b31d1d175485f8330df500044b155 : SUCCESS in 5m 59s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/8fb57080e9c645039ce27040f048e1cf : SUCCESS in 7m 02s","accounts_in_message":[],"_revision_number":3},{"id":"57d9ffe2110d288779c4f9c37aaab5bc6d7f0434","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33859,"name":"Yoon Soo Lim","display_name":"yoonsoolim","email":"msdbtjd123@naver.com","username":"yoonsoolim"},"date":"2023-05-24 10:48:35.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"90945c4270cf80b9e85e53237a8212dc9c0bbb7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31238,"name":"Youngju Lee","email":"leeyj7141@gmail.com","username":"leeyj7141"},"date":"2023-05-24 10:54:46.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2e94c5ed9061b15f1587a0c52f0b9ecb62f96d3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35982,"name":"Young-Tae Han","display_name":"young","email":"han0tae@gmail.com","username":"han0tae"},"date":"2023-05-24 10:55:10.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ddf0506476d6a8109a348530411f49028bf31641","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 11:06:43.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/919b62b1047a4c089a4755c7e225a69a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3ec7efbd8284c70a430f2539a414635 : SUCCESS in 4m 09s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/2f9094a326b5457997adf1deaae9490b : SUCCESS in 7m 52s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/e870b061eaff4df0b6f05a55b0f941e1 : SUCCESS in 5m 16s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/56e56ce1af1541b9b138f41254469513 : SUCCESS in 6m 42s","accounts_in_message":[],"_revision_number":6},{"id":"0ce879c3fe6bb6016842a774b96d476cd158679f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33859,"name":"Yoon Soo Lim","display_name":"yoonsoolim","email":"msdbtjd123@naver.com","username":"yoonsoolim"},"date":"2023-05-24 11:12:09.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"2d9d5e9c6f491f9dfb0c5d5bc0d6457d4bd26861","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31238,"name":"Youngju Lee","email":"leeyj7141@gmail.com","username":"leeyj7141"},"date":"2023-05-24 11:16:39.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"f66ef63898c5512cde315327aea00a62f36615bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 11:27:23.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c088f27531b493298d1bf1c6ebb8b7a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc42d3e484c4478da1f06dc5d0dc2ac1 : SUCCESS in 5m 22s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/25e6f4c1ea944477853f6b35f6334c6d : SUCCESS in 9m 26s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/6d310ebbdde44f44a544460a155e01e3 : SUCCESS in 6m 48s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/b5c48ab8474449bbaab40682f5aa6d20 : FAILURE in 7m 35s","accounts_in_message":[],"_revision_number":8},{"id":"e19cf47a3e72882f216812930c35d1bb5a708237","author":{"_account_id":35517,"name":"WonGyeongLee","email":"wongyeong.tech@gmail.com","username":"wongyeong"},"date":"2023-05-24 11:36:43.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"05156f30dc3651fd84902b3b2b4e6da4a7946fe9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 11:53:17.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c5545cca61647238769e9ffd7d6a8e4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fda966a273f14833a9fbd0f708279cf9 : SUCCESS in 4m 34s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/a41b90d4277f44b3ab7cf1b7392c10fa : SUCCESS in 10m 44s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/d01988216bce406bb9504b9d0bf93026 : SUCCESS in 6m 57s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/d8d83705bb7a4e77a9a98651fbfedb36 : FAILURE in 6m 55s","accounts_in_message":[],"_revision_number":8},{"id":"bb43ee7c791e5e19c1c04fb79ddc5ab242075d34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36001,"name":"Luther Lee","display_name":"Luther","email":"fashionday2@gmail.com","username":"fashionday2"},"date":"2023-05-24 12:03:15.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"afa42f50932c446917393c8837c5863d66eaf199","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33859,"name":"Yoon Soo Lim","display_name":"yoonsoolim","email":"msdbtjd123@naver.com","username":"yoonsoolim"},"date":"2023-05-24 12:04:46.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"e1d8bdd1b350a556088692c5a78cce0cc8ceb367","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33859,"name":"Yoon Soo Lim","display_name":"yoonsoolim","email":"msdbtjd123@naver.com","username":"yoonsoolim"},"date":"2023-05-24 12:05:55.000000000","message":"Patch Set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"e768a7cadf4744ec82b7978cf9ca7a3c7dd94acb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33859,"name":"Yoon Soo Lim","display_name":"yoonsoolim","email":"msdbtjd123@naver.com","username":"yoonsoolim"},"date":"2023-05-24 12:07:29.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"7cce830f131ab335ff78d5163e6812cd02125322","author":{"_account_id":33859,"name":"Yoon Soo Lim","display_name":"yoonsoolim","email":"msdbtjd123@naver.com","username":"yoonsoolim"},"date":"2023-05-24 12:08:05.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"f9773d27300495669a61587328fb49ba0b714b02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 12:17:40.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9d581fcc4894bbdb7bd5cf97ba9ee73\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd7753f5666a4056be755fbd34be21bd : SUCCESS in 4m 58s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/c106299ee1924e05b8f1d75ac58db43c : SUCCESS in 8m 26s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/10bec40348d541f2800c63e351173b3f : SUCCESS in 5m 28s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/50c0c1150cea43f18e35d16b6d73abdd : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":12},{"id":"53677777d3d613fe40a86466ba027f0747b7eb90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-05-25 20:07:18.000000000","message":"Patch Set 13: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"db93b9f72805a0eb1310644ce7df5c61ac75b168","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-05-25 20:08:18.000000000","message":"Patch Set 13: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"ea0bb7cde03eb19a7121c0cdd4b04afda40b68a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 20:27:23.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/940c0822d1b44037ad40a3b7992807b8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c56cb50d296463d8e959026233c9cc2 : SUCCESS in 4m 52s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/5d60b6a05a5744e7a3dae4b6b2481d42 : SUCCESS in 12m 10s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/8158036aca1b48558565920787d2f2cf : SUCCESS in 6m 44s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/5ccee094f2674e588a8dac3d0146d209 : SUCCESS in 8m 45s","accounts_in_message":[],"_revision_number":13},{"id":"2c0f61e951a3d91619ed5ffc17ef0159c05e7eb8","author":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"date":"2023-05-27 10:00:53.000000000","message":"Patch Set 13: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"01c761ab1f4b44523f5c7b5d5099a1f9360ad255","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-05-29 05:37:17.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"455cfff19e05ed557f34456b1c16fdb93337b797","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35057,"name":"JaeSeong-Shin","email":"rornfl916@gmail.com","username":"JaeSeong"},"date":"2023-05-30 11:27:36.000000000","message":"Patch Set 14: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"36e1d4b9d1c6d61d0e749e8b1c07f068adc7e90b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35057,"name":"JaeSeong-Shin","email":"rornfl916@gmail.com","username":"JaeSeong"},"date":"2023-05-30 11:28:04.000000000","message":"Patch Set 15: Commit message was updated.","accounts_in_message":[],"_revision_number":15},{"id":"45eb482c007d2cd1a2d5fc6ceb100ba71d0d5836","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 11:45:18.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d737d22756354011bbf5e60b7eb27071\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b44187b902344988a7b88ae151e66b6 : SUCCESS in 4m 27s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/cc8b3b7662654e0eb3fd399711842b18 : SUCCESS in 10m 17s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/63dc6de44f034fd28e2f2bb3fb729fb8 : SUCCESS in 8m 25s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/423e25108a764da6803a8d4dcf1d06fc : SUCCESS in 7m 16s","accounts_in_message":[],"_revision_number":15},{"id":"c9bfa989be3ff73a823873e9a7129415d8929cc5","author":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"date":"2023-06-08 16:23:28.000000000","message":"Patch Set 15: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"cb2b895cdabcd9ee7b9f2260bc3ff770d086a2e8","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-06-13 05:30:28.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"60600de6c220433a77eee866a31cc250ada03125","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33859,"name":"Yoon Soo Lim","display_name":"yoonsoolim","email":"msdbtjd123@naver.com","username":"yoonsoolim"},"date":"2023-06-27 02:10:02.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"cf63f9cb8488a217388112f996bbddacbf0d5d28","author":{"_account_id":33859,"name":"Yoon Soo Lim","display_name":"yoonsoolim","email":"msdbtjd123@naver.com","username":"yoonsoolim"},"date":"2023-06-27 02:11:28.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"a2fbfd76a249bf9463a392f8fccbfa3ed83e5811","author":{"_account_id":33859,"name":"Yoon Soo Lim","display_name":"yoonsoolim","email":"msdbtjd123@naver.com","username":"yoonsoolim"},"date":"2023-06-27 02:11:41.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"db086f83743785face96a3cd02cf4ad932282e08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 02:43:06.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac1b5af7e25a4b98bc1918ead1a25059\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/471450ac7c6f4887b3bbbc3a82545512 : SUCCESS in 6m 12s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/580d188cd1d049e9b0ec49dd95514ecd : SUCCESS in 11m 14s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/177c30ea9c8e4e0c985ea0e574fab5f8 : SUCCESS in 7m 42s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/dbbcf21668d8439b8982fc448ee9c50e : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":16},{"id":"bffa3995fb089a0cb7ef8c341c265fb2bd0b73fa","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-06-27 06:26:07.000000000","message":"Patch Set 16: Code-Review+2","accounts_in_message":[],"_revision_number":16},{"id":"40444787bc84d151395b61e8b49520bc63943de5","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-07-12 01:28:58.000000000","message":"Patch Set 16: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"897e75fdd671ad4a8f44bd046debce45c37f41ca","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-07-12 01:50:17.000000000","message":"Patch Set 16: Workflow+1","accounts_in_message":[],"_revision_number":16},{"id":"b51af9e884521a16c4635e041ebb3fa4d0d7de83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 01:50:43.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"38877a76159d67e2dae1a369397d5120b7e11548","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 01:58:20.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d670fb819b044d88af14a2d3bf28b9fd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d86d427e06444e2afab89bc8623330a : SUCCESS in 4m 27s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/7630e2b3ce954553a7a33cb3004f269f : SUCCESS in 7m 06s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/ad6466ef7d1a411cbcf05c6a4166fe6c : SUCCESS in 4m 24s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/1ca717f11963478ca9c6173d08063a3f : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":16},{"id":"9b97a4a03daf903323363f739c70f3ad619049f4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 01:58:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":16},{"id":"7f1de89dcfb1d1b13aa9dd81dc71afa7b8cad10a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 01:59:17.000000000","message":"Patch Set 16:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1df2efb545e04f5a9782ab6edeae12b1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b2cfdffe0f94d418d7448c96f91c7c1 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"dad25eeb8fc8919eda2f272d3b73930945993a1f","revisions":{"33e84dc3cb471a288bbcb42095ae155552eed4df":{"kind":"REWORK","_number":1,"created":"2023-05-17 12:07:35.000000000","uploader":{"_account_id":35517,"name":"WonGyeongLee","email":"wongyeong.tech@gmail.com","username":"wongyeong"},"ref":"refs/changes/64/883364/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/64/883364/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/1"}}},"commit":{"parents":[{"commit":"f395e0e2ec023884af67233dc13c33e7fd2d0c41","subject":"fix: fix MagicInput component","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f395e0e2ec023884af67233dc13c33e7fd2d0c41"}]}],"author":{"name":"WG19","email":"wongyeong3@gmail.com","date":"2023-05-17 12:01:45.000000000","tz":540},"committer":{"name":"wongyeong","email":"wongyeong.tech@gmail.com","date":"2023-05-17 12:07:04.000000000","tz":540},"subject":"Translated into Korean","message":"Translated into Korean\n\nChange-Id: I421469265e6f12f1583456bc20f8cafa7c48536e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/33e84dc3cb471a288bbcb42095ae155552eed4df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/33e84dc3cb471a288bbcb42095ae155552eed4df"}]},"branch":"refs/heads/master"},"559833547dd93f4defdcdedb32bf47c8b9ccd3aa":{"kind":"REWORK","_number":2,"created":"2023-05-17 12:24:18.000000000","uploader":{"_account_id":33859,"name":"Yoon Soo Lim","display_name":"yoonsoolim","email":"msdbtjd123@naver.com","username":"yoonsoolim"},"ref":"refs/changes/64/883364/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/64/883364/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/2"}}},"commit":{"parents":[{"commit":"f395e0e2ec023884af67233dc13c33e7fd2d0c41","subject":"fix: fix MagicInput component","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f395e0e2ec023884af67233dc13c33e7fd2d0c41"}]}],"author":{"name":"WG19","email":"wongyeong3@gmail.com","date":"2023-05-17 12:01:45.000000000","tz":540},"committer":{"name":"NHN","email":"msdbtjd123@naver.com","date":"2023-05-17 12:24:03.000000000","tz":540},"subject":"Translated into Korean","message":"Translated into Korean\n\nChange-Id: I421469265e6f12f1583456bc20f8cafa7c48536e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/559833547dd93f4defdcdedb32bf47c8b9ccd3aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/559833547dd93f4defdcdedb32bf47c8b9ccd3aa"}]},"branch":"refs/heads/master"},"7c5a84cb351b382a8d7a23f4c71a42b347d0d314":{"kind":"REWORK","_number":3,"created":"2023-05-24 02:21:35.000000000","uploader":{"_account_id":31238,"name":"Youngju Lee","email":"leeyj7141@gmail.com","username":"leeyj7141"},"ref":"refs/changes/64/883364/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/64/883364/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/3"}}},"commit":{"parents":[{"commit":"f395e0e2ec023884af67233dc13c33e7fd2d0c41","subject":"fix: fix MagicInput component","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f395e0e2ec023884af67233dc13c33e7fd2d0c41"}]}],"author":{"name":"WG19","email":"wongyeong3@gmail.com","date":"2023-05-17 12:01:45.000000000","tz":540},"committer":{"name":"youngju LEE","email":"leeyj7141@gmail.com","date":"2023-05-24 02:21:19.000000000","tz":540},"subject":"Translated into Korean","message":"Translated into Korean\n\nChange-Id: I421469265e6f12f1583456bc20f8cafa7c48536e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7c5a84cb351b382a8d7a23f4c71a42b347d0d314"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7c5a84cb351b382a8d7a23f4c71a42b347d0d314"}]},"branch":"refs/heads/master"},"431a74eeec638fce57b1322a40f875814c512e23":{"kind":"REWORK","_number":4,"created":"2023-05-24 10:48:35.000000000","uploader":{"_account_id":33859,"name":"Yoon Soo Lim","display_name":"yoonsoolim","email":"msdbtjd123@naver.com","username":"yoonsoolim"},"ref":"refs/changes/64/883364/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/64/883364/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/4"}}},"commit":{"parents":[{"commit":"f395e0e2ec023884af67233dc13c33e7fd2d0c41","subject":"fix: fix MagicInput component","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f395e0e2ec023884af67233dc13c33e7fd2d0c41"}]}],"author":{"name":"WG19","email":"wongyeong3@gmail.com","date":"2023-05-17 12:01:45.000000000","tz":540},"committer":{"name":"NHN","email":"msdbtjd123@naver.com","date":"2023-05-24 10:48:25.000000000","tz":540},"subject":"Translated into Korean","message":"Translated into Korean\n\nChange-Id: I421469265e6f12f1583456bc20f8cafa7c48536e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/431a74eeec638fce57b1322a40f875814c512e23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/431a74eeec638fce57b1322a40f875814c512e23"}]},"branch":"refs/heads/master"},"1cb2f840e6d85cb36c2bc3a0421acbecc12ca1af":{"kind":"REWORK","_number":5,"created":"2023-05-24 10:54:46.000000000","uploader":{"_account_id":31238,"name":"Youngju Lee","email":"leeyj7141@gmail.com","username":"leeyj7141"},"ref":"refs/changes/64/883364/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/64/883364/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/5"}}},"commit":{"parents":[{"commit":"f395e0e2ec023884af67233dc13c33e7fd2d0c41","subject":"fix: fix MagicInput component","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f395e0e2ec023884af67233dc13c33e7fd2d0c41"}]}],"author":{"name":"WG19","email":"wongyeong3@gmail.com","date":"2023-05-17 12:01:45.000000000","tz":540},"committer":{"name":"youngju.lee","email":"leeyj7141@gmail.com","date":"2023-05-24 10:54:17.000000000","tz":540},"subject":"Translated into Korean","message":"Translated into Korean\n\nChange-Id: I421469265e6f12f1583456bc20f8cafa7c48536e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1cb2f840e6d85cb36c2bc3a0421acbecc12ca1af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1cb2f840e6d85cb36c2bc3a0421acbecc12ca1af"}]},"branch":"refs/heads/master"},"b711aa415faaa2dec2b9ffd770c168a0f6e9a2da":{"kind":"REWORK","_number":6,"created":"2023-05-24 10:55:10.000000000","uploader":{"_account_id":35982,"name":"Young-Tae Han","display_name":"young","email":"han0tae@gmail.com","username":"han0tae"},"ref":"refs/changes/64/883364/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/64/883364/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/6"}}},"commit":{"parents":[{"commit":"f395e0e2ec023884af67233dc13c33e7fd2d0c41","subject":"fix: fix MagicInput component","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f395e0e2ec023884af67233dc13c33e7fd2d0c41"}]}],"author":{"name":"WG19","email":"wongyeong3@gmail.com","date":"2023-05-17 12:01:45.000000000","tz":540},"committer":{"name":"Young-Tae Han","email":"han0tae@gmail.com","date":"2023-05-24 10:54:49.000000000","tz":540},"subject":"Translated into Korean","message":"Translated into Korean\n\nChange-Id: I421469265e6f12f1583456bc20f8cafa7c48536e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b711aa415faaa2dec2b9ffd770c168a0f6e9a2da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b711aa415faaa2dec2b9ffd770c168a0f6e9a2da"}]},"branch":"refs/heads/master"},"4e8428239c3b82a9b9395c2d21d18e4ff6700595":{"kind":"REWORK","_number":7,"created":"2023-05-24 11:12:09.000000000","uploader":{"_account_id":33859,"name":"Yoon Soo Lim","display_name":"yoonsoolim","email":"msdbtjd123@naver.com","username":"yoonsoolim"},"ref":"refs/changes/64/883364/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/64/883364/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/7"}}},"commit":{"parents":[{"commit":"f395e0e2ec023884af67233dc13c33e7fd2d0c41","subject":"fix: fix MagicInput component","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f395e0e2ec023884af67233dc13c33e7fd2d0c41"}]}],"author":{"name":"WG19","email":"wongyeong3@gmail.com","date":"2023-05-17 12:01:45.000000000","tz":540},"committer":{"name":"NHN","email":"msdbtjd123@naver.com","date":"2023-05-24 11:11:51.000000000","tz":540},"subject":"Translated into Korean","message":"Translated into Korean\n\nChange-Id: I421469265e6f12f1583456bc20f8cafa7c48536e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4e8428239c3b82a9b9395c2d21d18e4ff6700595"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4e8428239c3b82a9b9395c2d21d18e4ff6700595"}]},"branch":"refs/heads/master"},"81288a5c0ffc87e6ebe4a366dfde86316a8ee217":{"kind":"REWORK","_number":8,"created":"2023-05-24 11:16:39.000000000","uploader":{"_account_id":31238,"name":"Youngju Lee","email":"leeyj7141@gmail.com","username":"leeyj7141"},"ref":"refs/changes/64/883364/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/64/883364/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/8"}}},"commit":{"parents":[{"commit":"f395e0e2ec023884af67233dc13c33e7fd2d0c41","subject":"fix: fix MagicInput component","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f395e0e2ec023884af67233dc13c33e7fd2d0c41"}]}],"author":{"name":"WG19","email":"wongyeong3@gmail.com","date":"2023-05-17 12:01:45.000000000","tz":540},"committer":{"name":"youngju.lee","email":"leeyj7141@gmail.com","date":"2023-05-24 11:16:24.000000000","tz":540},"subject":"Translated into Korean","message":"Translated into Korean\n\nChange-Id: I421469265e6f12f1583456bc20f8cafa7c48536e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/81288a5c0ffc87e6ebe4a366dfde86316a8ee217"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/81288a5c0ffc87e6ebe4a366dfde86316a8ee217"}]},"branch":"refs/heads/master"},"c38c1341a9dac060d78ec81c2d0cd6629112f1e5":{"kind":"REWORK","_number":9,"created":"2023-05-24 12:03:15.000000000","uploader":{"_account_id":36001,"name":"Luther Lee","display_name":"Luther","email":"fashionday2@gmail.com","username":"fashionday2"},"ref":"refs/changes/64/883364/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/64/883364/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/9"}}},"commit":{"parents":[{"commit":"f395e0e2ec023884af67233dc13c33e7fd2d0c41","subject":"fix: fix MagicInput component","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f395e0e2ec023884af67233dc13c33e7fd2d0c41"}]}],"author":{"name":"WG19","email":"wongyeong3@gmail.com","date":"2023-05-17 12:01:45.000000000","tz":540},"committer":{"name":"Luther Lee","email":"fashionday2@gmail.com","date":"2023-05-24 12:03:15.000000000","tz":0},"subject":"Translated into Korean","message":"Translated into Korean\n\nChange-Id: I421469265e6f12f1583456bc20f8cafa7c48536e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c38c1341a9dac060d78ec81c2d0cd6629112f1e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c38c1341a9dac060d78ec81c2d0cd6629112f1e5"}]},"branch":"refs/heads/master"},"895427db5f916efda5a8eb8ef97431d753ca43d5":{"kind":"REWORK","_number":10,"created":"2023-05-24 12:04:46.000000000","uploader":{"_account_id":33859,"name":"Yoon Soo Lim","display_name":"yoonsoolim","email":"msdbtjd123@naver.com","username":"yoonsoolim"},"ref":"refs/changes/64/883364/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/64/883364/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/10"}}},"commit":{"parents":[{"commit":"f395e0e2ec023884af67233dc13c33e7fd2d0c41","subject":"fix: fix MagicInput component","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f395e0e2ec023884af67233dc13c33e7fd2d0c41"}]}],"author":{"name":"WG19","email":"wongyeong3@gmail.com","date":"2023-05-17 12:01:45.000000000","tz":540},"committer":{"name":"Yoon Soo Lim","email":"msdbtjd123@naver.com","date":"2023-05-24 12:04:46.000000000","tz":0},"subject":"Translated into Korean","message":"Translated into Korean\n\nChange-Id: I421469265e6f12f1583456bc20f8cafa7c48536e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/895427db5f916efda5a8eb8ef97431d753ca43d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/895427db5f916efda5a8eb8ef97431d753ca43d5"}]},"branch":"refs/heads/master"},"e9af71a7f45d01e1e75ff47865fccf50d57577e2":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2023-05-24 12:05:55.000000000","uploader":{"_account_id":33859,"name":"Yoon Soo Lim","display_name":"yoonsoolim","email":"msdbtjd123@naver.com","username":"yoonsoolim"},"ref":"refs/changes/64/883364/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/64/883364/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/11"}}},"commit":{"parents":[{"commit":"f395e0e2ec023884af67233dc13c33e7fd2d0c41","subject":"fix: fix MagicInput component","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f395e0e2ec023884af67233dc13c33e7fd2d0c41"}]}],"author":{"name":"WG19","email":"wongyeong3@gmail.com","date":"2023-05-17 12:01:45.000000000","tz":540},"committer":{"name":"Yoon Soo Lim","email":"msdbtjd123@naver.com","date":"2023-05-24 12:05:55.000000000","tz":0},"subject":"Translated into Korean","message":"Translated into Korean\n\nCo-authored-by: Yoon Soo Lim \u003cmsdbtjd123@naver.com\u003e\nCo-authored-by: Youngju Lee \u003cleeyj7141@gmail.com\u003e\nCo-authored-by: Young-Tae Han \u003chan0tae@gmail.com\u003e\nCo-authored-by: Daewon Kim \u003cprudentcircle@outlook.com\u003e\nCo-authored-by: Jihye Park \u003cnina3909@naver.com\u003e\nCo-authored-by: hojin kim \u003cdncs0725@gmail.com\u003e\nCo-authored-by: Hocheol Shin \u003cshingoon7@gmail.com\u003e\nCo-authored-by: Seungho Lee \u003cfashionday2@gmail.com\u003e\n\nChange-Id: I421469265e6f12f1583456bc20f8cafa7c48536e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e9af71a7f45d01e1e75ff47865fccf50d57577e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e9af71a7f45d01e1e75ff47865fccf50d57577e2"}]},"branch":"refs/heads/master","description":"Edit commit message"},"986fc6c29dc117e321ba33d39ac7e9a27c222d0a":{"kind":"REWORK","_number":12,"created":"2023-05-24 12:07:29.000000000","uploader":{"_account_id":33859,"name":"Yoon Soo Lim","display_name":"yoonsoolim","email":"msdbtjd123@naver.com","username":"yoonsoolim"},"ref":"refs/changes/64/883364/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/64/883364/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/12"}}},"commit":{"parents":[{"commit":"f395e0e2ec023884af67233dc13c33e7fd2d0c41","subject":"fix: fix MagicInput component","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f395e0e2ec023884af67233dc13c33e7fd2d0c41"}]}],"author":{"name":"WG19","email":"wongyeong3@gmail.com","date":"2023-05-17 12:01:45.000000000","tz":540},"committer":{"name":"Yoon Soo Lim","email":"msdbtjd123@naver.com","date":"2023-05-24 12:07:28.000000000","tz":0},"subject":"Translated into Korean","message":"Translated into Korean\n\nCo-authored-by: Yoon Soo Lim \u003cmsdbtjd123@naver.com\u003e\nCo-authored-by: Youngju Lee \u003cleeyj7141@gmail.com\u003e\nCo-authored-by: Young-Tae Han \u003chan0tae@gmail.com\u003e\nCo-authored-by: Daewon Kim \u003cprudentcircle@outlook.com\u003e\nCo-authored-by: Jihye Park \u003cnina3909@naver.com\u003e\nCo-authored-by: hojin kim \u003cdncs0725@gmail.com\u003e\nCo-authored-by: Hocheol Shin \u003cshingoon7@gmail.com\u003e\nCo-authored-by: Seungho Lee \u003cfashionday2@gmail.com\u003e\n\nChange-Id: I421469265e6f12f1583456bc20f8cafa7c48536e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/986fc6c29dc117e321ba33d39ac7e9a27c222d0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/986fc6c29dc117e321ba33d39ac7e9a27c222d0a"}]},"branch":"refs/heads/master"},"b3eff40e26e8231857a97eca4b53e516ad26bb0c":{"kind":"NO_CODE_CHANGE","_number":13,"created":"2023-05-25 20:07:18.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/64/883364/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/64/883364/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/13"}}},"commit":{"parents":[{"commit":"f395e0e2ec023884af67233dc13c33e7fd2d0c41","subject":"fix: fix MagicInput component","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f395e0e2ec023884af67233dc13c33e7fd2d0c41"}]}],"author":{"name":"WG19","email":"wongyeong3@gmail.com","date":"2023-05-17 12:01:45.000000000","tz":540},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2023-05-25 20:07:18.000000000","tz":0},"subject":"feat: Update i18n Korean translation","message":"feat: Update i18n Korean translation\n\nNote: all of the co-authors collaborated the translation artifact together.\n\nCo-authored-by: Yoon Soo Lim \u003cmsdbtjd123@naver.com\u003e\nCo-authored-by: Youngju Lee \u003cleeyj7141@gmail.com\u003e\nCo-authored-by: Young-Tae Han \u003chan0tae@gmail.com\u003e\nCo-authored-by: Daewon Kim \u003cprudentcircle@outlook.com\u003e\nCo-authored-by: Jihye Park \u003cnina3909@naver.com\u003e\nCo-authored-by: hojin kim \u003cdncs0725@gmail.com\u003e\nCo-authored-by: Hocheol Shin \u003cshingoon7@gmail.com\u003e\nCo-authored-by: Seungho Lee \u003cfashionday2@gmail.com\u003e\n\nChange-Id: I421469265e6f12f1583456bc20f8cafa7c48536e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b3eff40e26e8231857a97eca4b53e516ad26bb0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b3eff40e26e8231857a97eca4b53e516ad26bb0c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"94ff007c6b12da53c1fe718f15af6b0f1a78e7d3":{"kind":"NO_CODE_CHANGE","_number":14,"created":"2023-05-30 11:27:36.000000000","uploader":{"_account_id":35057,"name":"JaeSeong-Shin","email":"rornfl916@gmail.com","username":"JaeSeong"},"ref":"refs/changes/64/883364/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/64/883364/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/14"}}},"commit":{"parents":[{"commit":"f395e0e2ec023884af67233dc13c33e7fd2d0c41","subject":"fix: fix MagicInput component","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f395e0e2ec023884af67233dc13c33e7fd2d0c41"}]}],"author":{"name":"WG19","email":"wongyeong3@gmail.com","date":"2023-05-17 12:01:45.000000000","tz":540},"committer":{"name":"JaeSeong-Shin","email":"rornfl916@gmail.com","date":"2023-05-30 11:27:36.000000000","tz":0},"subject":"feat: Update i18n Korean translation","message":"feat: Update i18n Korean translation\n\nNote: all of the co-authors collaborated the translation artifact together.\n\nCo-authored-by: Yoon Soo Lim \u003cmsdbtjd123@naver.com\u003e\nCo-authored-by: Youngju Lee \u003cleeyj7141@gmail.com\u003e\nCo-authored-by: Young-Tae Han \u003chan0tae@gmail.com\u003e\nCo-authored-by: Daewon Kim \u003cprudentcircle@outlook.com\u003e\nCo-authored-by: Jihye Park \u003cnina3909@naver.com\u003e\nCo-authored-by: hojin kim \u003cdncs0725@gmail.com\u003e\nCo-authored-by: Hocheol Shin \u003cshingoon7@gmail.com\u003e\nCo-authored-by: Seungho Lee \u003cfashionday2@gmail.com\u003e\nCo-authored-by: Jaeseong Shin \u003crornfl916@gmail.com\u003e\n\nChange-Id: I421469265e6f12f1583456bc20f8cafa7c48536e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/94ff007c6b12da53c1fe718f15af6b0f1a78e7d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/94ff007c6b12da53c1fe718f15af6b0f1a78e7d3"}]},"branch":"refs/heads/master","description":"Edit commit message"},"ac9dd5795d17472e9e3e2e8f83b1044098869e40":{"kind":"NO_CODE_CHANGE","_number":15,"created":"2023-05-30 11:28:04.000000000","uploader":{"_account_id":35057,"name":"JaeSeong-Shin","email":"rornfl916@gmail.com","username":"JaeSeong"},"ref":"refs/changes/64/883364/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/64/883364/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/15"}}},"commit":{"parents":[{"commit":"f395e0e2ec023884af67233dc13c33e7fd2d0c41","subject":"fix: fix MagicInput component","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f395e0e2ec023884af67233dc13c33e7fd2d0c41"}]}],"author":{"name":"WG19","email":"wongyeong3@gmail.com","date":"2023-05-17 12:01:45.000000000","tz":540},"committer":{"name":"JaeSeong-Shin","email":"rornfl916@gmail.com","date":"2023-05-30 11:28:04.000000000","tz":0},"subject":"feat: Update i18n Korean translation","message":"feat: Update i18n Korean translation\n\nNote: all of the co-authors collaborated the translation artifact together.\n\nCo-authored-by: Yoon Soo Lim \u003cmsdbtjd123@naver.com\u003e\nCo-authored-by: Youngju Lee \u003cleeyj7141@gmail.com\u003e\nCo-authored-by: Young-Tae Han \u003chan0tae@gmail.com\u003e\nCo-authored-by: Daewon Kim \u003cprudentcircle@outlook.com\u003e\nCo-authored-by: Jihye Park \u003cnina3909@naver.com\u003e\nCo-authored-by: hojin kim \u003cdncs0725@gmail.com\u003e\nCo-authored-by: Hocheol Shin \u003cshingoon7@gmail.com\u003e\nCo-authored-by: Seungho Lee \u003cfashionday2@gmail.com\u003e\nCo-authored-by: Jaeseong-Shin \u003crornfl916@gmail.com\u003e\n\nChange-Id: I421469265e6f12f1583456bc20f8cafa7c48536e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ac9dd5795d17472e9e3e2e8f83b1044098869e40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ac9dd5795d17472e9e3e2e8f83b1044098869e40"}]},"branch":"refs/heads/master","description":"Edit commit message"},"dad25eeb8fc8919eda2f272d3b73930945993a1f":{"kind":"REWORK","_number":16,"created":"2023-06-27 02:10:02.000000000","uploader":{"_account_id":33859,"name":"Yoon Soo Lim","display_name":"yoonsoolim","email":"msdbtjd123@naver.com","username":"yoonsoolim"},"ref":"refs/changes/64/883364/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/64/883364/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/64/883364/16"}}},"commit":{"parents":[{"commit":"d07227822f26164ba8dee5798a0f675d09995079","subject":"feat: add jwt expire check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d07227822f26164ba8dee5798a0f675d09995079"}]}],"author":{"name":"WG19","email":"wongyeong3@gmail.com","date":"2023-05-17 12:01:45.000000000","tz":540},"committer":{"name":"NHN","email":"msdbtjd123@naver.com","date":"2023-06-27 02:09:48.000000000","tz":540},"subject":"feat: Update i18n Korean translation","message":"feat: Update i18n Korean translation\n\nNote: all of the co-authors collaborated the translation artifact together.\n\nCo-authored-by: Yoon Soo Lim \u003cmsdbtjd123@naver.com\u003e\nCo-authored-by: Youngju Lee \u003cleeyj7141@gmail.com\u003e\nCo-authored-by: Young-Tae Han \u003chan0tae@gmail.com\u003e\nCo-authored-by: Daewon Kim \u003cprudentcircle@outlook.com\u003e\nCo-authored-by: Jihye Park \u003cnina3909@naver.com\u003e\nCo-authored-by: hojin kim \u003cdncs0725@gmail.com\u003e\nCo-authored-by: Hocheol Shin \u003cshingoon7@gmail.com\u003e\nCo-authored-by: Seungho Lee \u003cfashionday2@gmail.com\u003e\nCo-authored-by: Jaeseong-Shin \u003crornfl916@gmail.com\u003e\n\nChange-Id: I421469265e6f12f1583456bc20f8cafa7c48536e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/dad25eeb8fc8919eda2f272d3b73930945993a1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/dad25eeb8fc8919eda2f272d3b73930945993a1f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
