)]}'
{"id":"openstack%2Fskyline-console~884010","triplet_id":"openstack%2Fskyline-console~master~I9b82c4624826854d40d11da4e6ce4a9137b86085","project":"openstack/skyline-console","branch":"master","topic":"color","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-05-26 11:00:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9b82c4624826854d40d11da4e6ce4a9137b86085","subject":"feat: update color variables","status":"MERGED","created":"2023-05-24 08:49:26.000000000","updated":"2023-05-26 11:01:25.000000000","submitted":"2023-05-26 11:00:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"884010-color","meta_rev_id":"68c319837173c7f0154f6fe185e20c6a25f19fc7","_number":884010,"virtual_id_number":884010,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-26 11:00:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-05-26 09:48:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-05-26 09:48:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-24 10:38:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-26 09:48:08.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"07fb00624d6ad9de384d2ce95536bc4eaa20b7b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-05-24 08:49:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58cc7557ad81321253795f9054a7e625f5ce8901","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 10:38:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c3d303b329504ae7a223290e38b3b640\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9adb48a54214153bd29cbc2336d0916 : SUCCESS in 4m 35s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/a29be81eb001419393bdc3670cd81f9c : SUCCESS in 7m 25s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/a67a1bc568314a4c9f96f0273f6e05f5 : SUCCESS in 5m 39s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/f1c8ce1eab764ba395fdd04494882c3d : SUCCESS in 7m 28s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/f27476e5a2604a8c86fb82dbcc57c440 : FAILURE in 1h 43m 59s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/83826d8cfe28401aa0eff993629937d9 : FAILURE in 1h 16m 19s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/e282ad6cc8b1426fa6b11974c9757adf : SUCCESS in 49m 04s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/e776376a50884a3281f33854544ad28b : FAILURE in 52m 30s","accounts_in_message":[],"_revision_number":1},{"id":"bae9835bc8e57267da75e66a87f7f31bc364e4ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-05-25 08:45:21.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"810f7d35b9de3eb4006682d1db7b83fc962d2261","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 10:11:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3293ba31936f47e1aa35e6fd77ab2cf9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04d112129efa4d4895a9c1910edb8521 : SUCCESS in 4m 37s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/08a166b5ea62470b8bb3e058fc975974 : SUCCESS in 6m 50s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/73f1c8e8b66d439c9bc51d67b3ced27a : SUCCESS in 10m 28s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/1425f5e7be7e4a9b98981b21396afc8d : SUCCESS in 7m 09s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/8497909241b943a296952145bd3ad91f : FAILURE in 1h 24m 09s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/2eda5538b9274a35ae66a15198b7b91c : SUCCESS in 55m 33s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/cc748b5e1d2243f4aa958dd3eb04f6db : SUCCESS in 57m 38s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/bbd259a56cf64dbdb9dc17d38ba4816f : SUCCESS in 56m 02s","accounts_in_message":[],"_revision_number":2},{"id":"89be77aef7a471addf646215a03b42e3ad64b391","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-05-26 00:57:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"253c70f29f28b9cc2551373fc8d7cee9fc02e30b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-26 02:34:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ed8831cee6c488fbea7b15b15879ceb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9734d06c78b8411da0636286be581b76 : SUCCESS in 4m 36s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/3328777ede2c4a3dabd0b19be88b4677 : SUCCESS in 10m 07s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/cb534403c1ab4b14b2ebc4ae0185ecd9 : SUCCESS in 5m 53s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/d222eb3b190a4aa8b9e8395f6b7c6705 : SUCCESS in 7m 52s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/8d502169df24470a977e460e523c916f : SUCCESS in 1h 29m 20s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/7001c7241b124e239375d0b92e79da24 : SUCCESS in 59m 25s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/c8614def5ee14c5b9b6c5b299fbdacf0 : SUCCESS in 48m 42s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/a3be1576b47c4cbab84c94ee68ab352b : SUCCESS in 54m 41s","accounts_in_message":[],"_revision_number":2},{"id":"3f0cdca9a52c8c0b4b0350e9bbdd91b008d6121b","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-05-26 09:48:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8d7e533500ebf3cff32c9b8e02018835ed4bad08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-26 09:48:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fe484a0af5046be883ba6be900333e8f99f5a190","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-26 11:00:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/452fd1fd082a409195c580050a9e2159\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c08ae3287de4d19906984127199a3f9 : SUCCESS in 5m 36s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/40b8b41379244a3cb30806d47820be48 : SUCCESS in 7m 13s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/74dc0eaa8c6446159ef0333bb5ac7242 : SUCCESS in 5m 17s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/22607f0f5dce4544b33239cd0844d345 : SUCCESS in 7m 30s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/8ba4bbde8286422888e624df9f576511 : SUCCESS in 1h 10m 25s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/20a31b121eef45bc850f5f04248841ba : SUCCESS in 53m 09s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/974f503089c44079a3f86eeefe5b3a48 : SUCCESS in 49m 53s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/957f565396fe4944bd23863a29c80123 : SUCCESS in 47m 47s","accounts_in_message":[],"_revision_number":2},{"id":"f5c0939841315648eb411d843097183746a33cb8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-26 11:00:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"68c319837173c7f0154f6fe185e20c6a25f19fc7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-26 11:01:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10f3963619124bcf99ee94c80b78e6d8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e90d4d8b3b94557b2a4432bf00a5ea9 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ec59419f4fd698523aa9b8718b8bd5d490b2464d","revisions":{"33fb39b1431d2949534b9761a124adba04b11a64":{"kind":"REWORK","_number":1,"created":"2023-05-24 08:49:26.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/10/884010/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/10/884010/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/10/884010/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/10/884010/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/10/884010/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/10/884010/1"}}},"commit":{"parents":[{"commit":"f395e0e2ec023884af67233dc13c33e7fd2d0c41","subject":"fix: fix MagicInput component","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f395e0e2ec023884af67233dc13c33e7fd2d0c41"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-05-24 08:48:56.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-05-24 08:48:56.000000000","tz":480},"subject":"feat: update color variables","message":"feat: update color variables\n\nUse color variables to set color in less and jsx, this makes it easy to change the theme for all pages.\n\nChange-Id: I9b82c4624826854d40d11da4e6ce4a9137b86085\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/33fb39b1431d2949534b9761a124adba04b11a64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/33fb39b1431d2949534b9761a124adba04b11a64"}]},"branch":"refs/heads/master"},"ec59419f4fd698523aa9b8718b8bd5d490b2464d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-05-25 08:45:21.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/10/884010/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/10/884010/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/10/884010/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/10/884010/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/10/884010/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/10/884010/2"}}},"commit":{"parents":[{"commit":"8ecf24e69f217d0bff20e5a5ba50785df8d05e11","subject":"Merge \"feat: update the cirros version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8ecf24e69f217d0bff20e5a5ba50785df8d05e11"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-05-24 08:48:56.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-05-25 08:44:59.000000000","tz":480},"subject":"feat: update color variables","message":"feat: update color variables\n\nUse color variables to set color in less and jsx, this makes it easy to change the theme for all pages.\n\nChange-Id: I9b82c4624826854d40d11da4e6ce4a9137b86085\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ec59419f4fd698523aa9b8718b8bd5d490b2464d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ec59419f4fd698523aa9b8718b8bd5d490b2464d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
