)]}'
{"id":"openstack%2Fskyline-console~884306","triplet_id":"openstack%2Fskyline-console~master~I0bb35d350f21dbae3fd66922b21522a579eb8967","project":"openstack/skyline-console","branch":"master","topic":"cirros","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-05-25 06:58:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0bb35d350f21dbae3fd66922b21522a579eb8967","subject":"feat: update the cirros version","status":"MERGED","created":"2023-05-25 01:31:35.000000000","updated":"2023-05-25 06:59:03.000000000","submitted":"2023-05-25 06:58:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"884306-cirros","meta_rev_id":"ac9c4853dbeb7d926a5d1bac9f0c087182f0fdcb","_number":884306,"virtual_id_number":884306,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-25 06:58:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-05-25 05:25:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-05-25 05:25:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-25 03:09:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-25 05:25:50.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"d10f290d05b13b4a9f8fbefddf5650ece777648f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-05-25 01:31:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73720b9fe32e82978a9fa40783997ab897ce9fa1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 03:09:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7356eddcab8a4711a8542d1e2779512b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/511c5fcea7c449438ed9200dc95bdcc3 : SUCCESS in 4m 51s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/e52957b7fafd45cab18043fd1f47a310 : SUCCESS in 6m 22s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/00cab6a43c8f447dbdbf9d86d57821ff : SUCCESS in 4m 20s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/6fbf1a92e5a843f9a154c3926f7612ec : SUCCESS in 4m 59s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/60bd7359849d4dabb408439ac20c3eaa : SUCCESS in 1h 05m 07s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/580c8b45545943308893b30bada1b33d : SUCCESS in 45m 42s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/2b9f7a2bd61f49d9b39872c00f3c15f3 : SUCCESS in 1h 03m 43s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/0a0aa869489e414b8907ac67be57f2cf : FAILURE in 1h 21m 54s","accounts_in_message":[],"_revision_number":1},{"id":"f8bfef24fb411ba5f6d3e37cfd40c2f2493b4722","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-05-25 04:12:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6a49a43e88d57954f439234ce3e12cc933c7c5b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 05:21:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afbd7d73e14349a38746c7fb4372560b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fd60ca5f69541a480c5c416dc4b3285 : SUCCESS in 4m 29s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/1803c3e45fa146a68215e87166e30306 : SUCCESS in 10m 21s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/6f6b9096ba15446f93b7d2b4d0bedf4c : SUCCESS in 6m 00s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/d5e332a6aed04deaa6e8bd859ca832a9 : SUCCESS in 4m 48s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/63e85dee9c26449f846bfcdf11687dbc : SUCCESS in 1h 06m 21s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/f4b1119196b14282afc9a8224bc9e055 : SUCCESS in 51m 54s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/6145aa6060c947009e483c547b050a0d : SUCCESS in 40m 52s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/90d9e199e7404e2b9d4c8658bd5726cb : SUCCESS in 40m 38s","accounts_in_message":[],"_revision_number":1},{"id":"9121bdae606ca1383bb6b4c0c78b6f186afa5f07","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-05-25 05:25:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f39835f8b3550ef58582b7371549cec7fa22a81c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 05:26:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"23d43f0dbc625a4159812114d2cd9f19ba94f8c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 06:58:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6149b211465458ab278d181225e5663\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e86e41b8ee494feba8335451a3b6361f : SUCCESS in 3m 53s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/2ddbf675ce1a4e1b97179eeb9ce7caf9 : SUCCESS in 8m 42s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/797b7029211f4039b2fe8e70526c124f : SUCCESS in 7m 41s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/0a97812731724307aaf4b98741a42224 : SUCCESS in 7m 47s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/a7d6acf8fd164b7cbf6fc9e841036200 : SUCCESS in 1h 27m 02s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/9192c3a8d72e4489a22c9b70a78b026e : SUCCESS in 41m 53s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/6bfe590679a542538433ce6216cf49cb : SUCCESS in 47m 48s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/5e745eb1666b459f9b49592a0ba15e80 : SUCCESS in 38m 05s","accounts_in_message":[],"_revision_number":1},{"id":"20266081c8609ee635f698df30a17c85e6231005","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 06:58:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ac9c4853dbeb7d926a5d1bac9f0c087182f0fdcb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 06:59:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/340d4d9802f641fdab9c957cf7439ff2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30df0ea405da46dc87b4a9be41c94db8 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"36a69b0a1f6a67400496d8af00b8b0c433b6fde8","revisions":{"36a69b0a1f6a67400496d8af00b8b0c433b6fde8":{"kind":"REWORK","_number":1,"created":"2023-05-25 01:31:35.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/06/884306/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/06/884306/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/06/884306/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/06/884306/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/06/884306/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/06/884306/1"}}},"commit":{"parents":[{"commit":"c73fd49f225745a68f428cb7f816ff35504b8f06","subject":"Merge \"Add Designate (DNS) UI to Skyline-Console\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c73fd49f225745a68f428cb7f816ff35504b8f06"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-05-25 01:30:48.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-05-25 01:30:48.000000000","tz":480},"subject":"feat: update the cirros version","message":"feat: update the cirros version\n\nUpdate the default cirros version used by e2e according to the devstack\n\nChange-Id: I0bb35d350f21dbae3fd66922b21522a579eb8967\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/36a69b0a1f6a67400496d8af00b8b0c433b6fde8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/36a69b0a1f6a67400496d8af00b8b0c433b6fde8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
