)]}'
{"id":"openstack%2Fskyline-console~885737","triplet_id":"openstack%2Fskyline-console~master~I3d407535cc389312c4c314e2a2f10cad1979679a","project":"openstack/skyline-console","branch":"master","topic":"project","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-06-09 12:46:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3d407535cc389312c4c314e2a2f10cad1979679a","subject":"feat: add domain link in projects page","status":"MERGED","created":"2023-06-09 09:31:34.000000000","updated":"2023-06-09 12:47:29.000000000","submitted":"2023-06-09 12:46:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"885737-project","meta_rev_id":"c1b3c46e5adaf449d7452af6ba024e8c450847b7","_number":885737,"virtual_id_number":885737,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-09 12:46:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-06-09 11:57:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-06-09 11:57:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-09 10:19:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-09 11:57:52.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"05864895db49a631792b5ff1f59746bd93b70bb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-06-09 09:31:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b6a54ff9b4b86aa6f2968fad72515233f46fce0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-09 10:19:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8d7290325cf46058be6dc5160aa14e9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71b0fbb4b180459ab4186aa7639430b1 : SUCCESS in 4m 09s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/ffce9e4174744d8196f38075f8ee7b0c : SUCCESS in 10m 14s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/dbbbf8459a274e83962a0a40adc36660 : SUCCESS in 7m 49s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/47e657a3a3d447e2ac394c74c89d2cf3 : SUCCESS in 7m 10s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/0b205ec755e54fb4b5f42f14575b1230 : SUCCESS in 46m 51s","accounts_in_message":[],"_revision_number":1},{"id":"f3188a3b216013e45d38ac8fa8581358bf5aa8f6","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-06-09 11:57:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c02b1052472c1d7c3afeb6338a293bf353ef66c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-09 11:58:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6f78bf4c29524c9590ebeacda9bc119d8d4ef7d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-09 12:46:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d525133e572f465ca7c554ee3371233b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/742d4fdcf6114584a26f2d3a22636c9d : SUCCESS in 5m 33s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/5adabe7bc1914f6f8e62e95e52f8d1b7 : SUCCESS in 9m 57s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/9e8ae96cb62748279d3e5d29b26e0cb2 : SUCCESS in 7m 00s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/02d2f9f4646b4672a72429cfa672322a : SUCCESS in 6m 38s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/1d143fb6e48447c891c27d1e393c4d24 : SUCCESS in 47m 00s","accounts_in_message":[],"_revision_number":1},{"id":"3e55304fb0111a90b833ea8a705720823a4039f0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-09 12:46:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c1b3c46e5adaf449d7452af6ba024e8c450847b7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-09 12:47:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03adb4dca31e49e9b96a6a7b8c4ded60\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f790380194d4a39b50d3a205fb2d004 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ee25e44c4eba0a9ef66abf500b87f37d30bdc047","revisions":{"ee25e44c4eba0a9ef66abf500b87f37d30bdc047":{"kind":"REWORK","_number":1,"created":"2023-06-09 09:31:34.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/37/885737/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/37/885737/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/37/885737/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/37/885737/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/37/885737/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/37/885737/1"}}},"commit":{"parents":[{"commit":"861876cddfa3ece91f1513aa3385e5c659d33c7e","subject":"fix: fix the virtual resource in administrator","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/861876cddfa3ece91f1513aa3385e5c659d33c7e"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-06-09 09:31:13.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-06-09 09:31:13.000000000","tz":480},"subject":"feat: add domain link in projects page","message":"feat: add domain link in projects page\n\nAdd domain id, domain detail link for the domain column in the project list page\n\nChange-Id: I3d407535cc389312c4c314e2a2f10cad1979679a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ee25e44c4eba0a9ef66abf500b87f37d30bdc047"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ee25e44c4eba0a9ef66abf500b87f37d30bdc047"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
