)]}'
{"id":"openstack%2Fskyline-console~885930","triplet_id":"openstack%2Fskyline-console~master~Ia909771d38477bf726cec25b1ad46b7780b64244","project":"openstack/skyline-console","branch":"master","topic":"bug/2023608","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-06-13 06:10:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia909771d38477bf726cec25b1ad46b7780b64244","subject":"fix: fix edit fip","status":"MERGED","created":"2023-06-13 02:09:37.000000000","updated":"2023-06-13 06:11:58.000000000","submitted":"2023-06-13 06:10:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"885930-bug/2023608","meta_rev_id":"939977f98392b76db4a5efbf224fcdfded38d156","_number":885930,"virtual_id_number":885930,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-13 06:10:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-06-13 05:28:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":1,"date":"2023-06-13 03:58:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-06-13 05:28:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-13 03:21:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-13 03:58:39.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"state":"REVIEWER"},{"updated":"2023-06-13 05:28:35.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"2c6638f7275055b94d7a3e242e469bae6e6ab3b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-06-13 02:09:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fdcfb31254e0e60b18a8e3a03a6588c674eb00f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 03:21:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc2bd83ea8044b3ab578989a11c23b56\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a030a53cbc8341c6b5c4336bba1c774f : SUCCESS in 5m 18s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/c94d53a48dc740e0b7d313c5202de3fa : SUCCESS in 9m 14s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/fa425a470b554bebb88d08681ee8c73f : SUCCESS in 6m 26s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/60561b350f874a4ea3ae21fdece59d75 : SUCCESS in 6m 57s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/05b15cb36d844768b27f6acc63789461 : SUCCESS in 59m 48s","accounts_in_message":[],"_revision_number":1},{"id":"48f069dcefb66c63edae4bb355635294bd6e8b8c","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-06-13 03:58:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f80cef3222e87056310fee99d1250d30e75cb6b6","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-06-13 05:28:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0240dfc36e24cc1906dd81dc2f18dd6b216ccc9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 05:29:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0fadc639222e541b6e4620892db7aaf1423779f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 06:10:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a272ce2b0b24680a93c4e0fc3d09ffa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8edbf1ebd294a889750760a9fa86309 : SUCCESS in 4m 39s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/40e380be29144293b48e8aba9879358b : SUCCESS in 10m 56s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/e710ca4722b54a19a33143cb6485561d : SUCCESS in 7m 05s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/6d262fa66f034b3e970057613c8e666c : SUCCESS in 4m 36s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/67e0ead0dd64402ba1a3f8f2b18039b1 : SUCCESS in 41m 48s","accounts_in_message":[],"_revision_number":1},{"id":"b9fcfc61f6d2317b2bb0979b9f1e7f353b2ac28d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 06:10:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"939977f98392b76db4a5efbf224fcdfded38d156","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 06:11:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5d200ce725a4af2a90116109d27217c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/982e2f6eee364017885405ea3e9a45e7 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a0dcb587c5426bf36cf3eee24b520df32b8b60de","revisions":{"a0dcb587c5426bf36cf3eee24b520df32b8b60de":{"kind":"REWORK","_number":1,"created":"2023-06-13 02:09:37.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/30/885930/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/30/885930/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/30/885930/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/30/885930/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/30/885930/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/30/885930/1"}}},"commit":{"parents":[{"commit":"ee25e44c4eba0a9ef66abf500b87f37d30bdc047","subject":"feat: add domain link in projects page","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ee25e44c4eba0a9ef66abf500b87f37d30bdc047"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-06-13 02:09:11.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-06-13 02:09:11.000000000","tz":480},"subject":"fix: fix edit fip","message":"fix: fix edit fip\n\nfix edit fip when the qos policy service is disabled\n\nCloses-Bug: #2023608\nChange-Id: Ia909771d38477bf726cec25b1ad46b7780b64244\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a0dcb587c5426bf36cf3eee24b520df32b8b60de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a0dcb587c5426bf36cf3eee24b520df32b8b60de"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
