)]}'
{"id":"openstack%2Fskyline-console~885953","triplet_id":"openstack%2Fskyline-console~master~I7859b91ced6549c26655b1624f28bb19427194e6","project":"openstack/skyline-console","branch":"master","topic":"identity","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-06-13 11:29:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7859b91ced6549c26655b1624f28bb19427194e6","subject":"feat: support domain name fuzzy search","status":"MERGED","created":"2023-06-13 09:32:38.000000000","updated":"2023-06-13 11:30:05.000000000","submitted":"2023-06-13 11:29:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"885953-identity","meta_rev_id":"f1f816ab6dd4d7e940f583cec3e22897d74b7a17","_number":885953,"virtual_id_number":885953,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-13 11:29:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-06-13 10:41:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-06-13 10:41:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-13 10:31:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-13 10:41:20.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"d6fc9d60f9b09db4fadb3473fc7f8b16f0ce9a9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-06-13 09:32:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ebaaea3fe4adbdc2d8489bd8eb52cfd4c60dba3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 10:31:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef53de7b6c2545e2989da8b4c51c2bc4\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9eb1067cdb8d4253b0995551d3037ce9 : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7af03e7b368d4b008a36af93ce0d63cc : SUCCESS in 5m 33s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/f44df45646524a47bbba3ad6195961cd : SUCCESS in 10m 32s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/9dc526964be9472f8fe1edfa9784c393 : SUCCESS in 7m 35s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/d8c40d8718f64cd78def3eb5992babb6 : SUCCESS in 6m 38s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/3357e9550e3c4427b4b9ddb5835225fc : SUCCESS in 53m 45s","accounts_in_message":[],"_revision_number":1},{"id":"58c88ea9683cc9c4312b478ee30b3bdc754d2dd6","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-06-13 10:41:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"800f5ca049f59999aec59d363933a7d652a02d8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 10:41:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"51cffb03d89454495d64be6423577b4eb052b95a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 11:29:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ccf95e049764ac88952ed3d018b968a\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68873387767f498b8a7a295db623e410 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46cd27298eb6404e9093c8915bfef661 : SUCCESS in 5m 06s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/ab988db6342141d5a18b4ab7e2de87fb : SUCCESS in 7m 23s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/40b91b69e73e4f9983ac39a96768f2bf : SUCCESS in 5m 47s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/8b386adc021a4577bc4915ee43aec8dd : SUCCESS in 5m 34s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/a14c27269eab48bf984af54800e2d9d4 : SUCCESS in 46m 45s","accounts_in_message":[],"_revision_number":1},{"id":"853c3af7199adee96fe0166d6b50e99dae75f73a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 11:29:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f1f816ab6dd4d7e940f583cec3e22897d74b7a17","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 11:30:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4437fd4f1fae423ebf827a06dd9e82c3\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/912ed7a45ea6488c907597e9ee802875 : SUCCESS in 38s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f77c4f518f8c4d13b4ceb42168f6a8c3 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2d6f72004126823bf4d188bf7c6f385ad7a4e620","revisions":{"2d6f72004126823bf4d188bf7c6f385ad7a4e620":{"kind":"REWORK","_number":1,"created":"2023-06-13 09:32:38.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/53/885953/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/53/885953/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/53/885953/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/53/885953/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/53/885953/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/53/885953/1"}}},"commit":{"parents":[{"commit":"eedea3eccd2f300bfe7cce5270db9887d4150b6f","subject":"fix: update create metadata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/eedea3eccd2f300bfe7cce5270db9887d4150b6f"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-06-13 09:32:18.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-06-13 09:32:18.000000000","tz":480},"subject":"feat: support domain name fuzzy search","message":"feat: support domain name fuzzy search\n\nSupport domain name fuzzy search for users, user groups and projects\n\nChange-Id: I7859b91ced6549c26655b1624f28bb19427194e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2d6f72004126823bf4d188bf7c6f385ad7a4e620"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2d6f72004126823bf4d188bf7c6f385ad7a4e620"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
