)]}'
{"id":"openstack%2Fskyline-console~886045","triplet_id":"openstack%2Fskyline-console~master~I4a33649d3dbf630e36bcdee71462b4796e17f714","project":"openstack/skyline-console","branch":"master","topic":"jwt","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-06-15 06:26:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4a33649d3dbf630e36bcdee71462b4796e17f714","subject":"feat: add jwt expire check","status":"MERGED","created":"2023-06-14 08:34:52.000000000","updated":"2023-06-15 06:27:31.000000000","submitted":"2023-06-15 06:26:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886045-jwt","meta_rev_id":"628930db079b78c1cd29e2a3b2fe4a8731faa8a0","_number":886045,"virtual_id_number":886045,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-15 06:26:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-06-15 05:24:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":1,"date":"2023-06-15 04:26:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-06-15 05:24:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-14 10:49:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-15 04:26:36.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"state":"REVIEWER"},{"updated":"2023-06-15 05:24:31.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"da205d7c5e952f0df5e0798247924aeae3e19e40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-06-14 08:34:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c229acf7751b4e8f608d649762a0ae88482cb6ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 10:49:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d8a5e4355b8b4c43951bcdd39eae60ea\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa416c200a21404a9cf9582100008ef5 : SUCCESS in 2m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45a17cade5ca4b058ebe5c46a9763b89 : SUCCESS in 4m 41s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/f06d9479493f472ba188a5d15d016281 : SUCCESS in 7m 30s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/b68a962e71594463bd8599c4174c729c : SUCCESS in 4m 52s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/1be06b02538a456c9edb9ce8bf54de4c : SUCCESS in 5m 27s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/d5e9470f15f8489ab95066988f5261c6 : FAILURE in 2h 09m 49s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/d65d6ab41595456faa25189e2678ec3e : FAILURE in 1h 28m 54s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/fac8e198eacc45928ca3891999650def : FAILURE in 2h 09m 05s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/c7e3c297da294ae0821c25199c80be42 : FAILURE in 1h 21m 24s","accounts_in_message":[],"_revision_number":1},{"id":"bea4a9aab47199fd79f23f5725d3305c032a613b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-06-15 01:12:53.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6c27a45675d05c7305a0ba1c177ce5067fdd06d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-06-15 02:44:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d3adb7f6c63e49c0c56f2929517b619cdf677c37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 04:22:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53059d410963482cbd008130b150eead\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1cf0b33768134b23a3f7ffd697ee4201 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e318783c25134b739bf2251bd27a1ccf : SUCCESS in 15m 25s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/af955bd024e14e468e4740c50213f1a3 : SUCCESS in 7m 54s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/840951baace24d5aac964cb6bec70a47 : SUCCESS in 5m 52s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/59cc29e70bcc4fe99e9ab5ac374421f8 : SUCCESS in 6m 40s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/24d63a74a50d44858ab05cf52d4874f8 : SUCCESS in 1h 17m 44s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/097e6ce0002b41989d96153b4f73f241 : SUCCESS in 1h 21m 31s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/2c876618eab44d9d81762d70bf602e08 : SUCCESS in 51m 21s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/8e28fb94af9e480e8ac322f9ed668e2a : SUCCESS in 48m 52s","accounts_in_message":[],"_revision_number":3},{"id":"47c49f2f7ae0c26bd8c7f2ec03310058823eef9c","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-06-15 04:26:36.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"4097957a40c7825a9af70ea1e37b0bfedbaa2195","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-06-15 05:24:31.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c62c4bc3e545c21a9f1063da16604941e5c71e39","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-06-15 05:24:40.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6402cdf2e4bcee599851c7a820ac49f3adbf6b69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 05:25:16.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d7e65234735126e975322029789492cf104ef961","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 06:26:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76bdf88e57e842c49cb07015a74943ba\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e63eec9e609c40e5a5b518afa9983253 : SUCCESS in 2m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dcf81cf9fcc4b32a53bb95c3708f8c6 : SUCCESS in 4m 31s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/036e6751b9f54859b312a66e621055e0 : SUCCESS in 6m 57s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/c028384cea6b429bbdfae586cadc187a : SUCCESS in 4m 42s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/f94857c8369841af87df2fd1fa4b5994 : SUCCESS in 5m 04s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/8d0bed415f604b54b7ef05f2edc7017d : SUCCESS in 1h 00m 49s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/49b05220c7914cff868287dee9dfc07c : SUCCESS in 45m 23s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/e57dfab284814b728e32ccc3360cdb00 : SUCCESS in 43m 01s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/9eb43d4c85f147f097774933a600c5b0 : SUCCESS in 40m 19s","accounts_in_message":[],"_revision_number":3},{"id":"80c8a9a050873d48f86a6a8f447dd4a41a8e2921","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 06:26:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"628930db079b78c1cd29e2a3b2fe4a8731faa8a0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 06:27:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/779196f3855942f48fd7d1ae639522b5\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13abf9b042b14f9392703010be1d937c : SUCCESS in 33s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0d56d2d7cac4acc9008a42cb747e3aa : SUCCESS in 35s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d07227822f26164ba8dee5798a0f675d09995079","revisions":{"4a6c5132e02848b2ae4ccea84bc493c7a0634eb2":{"kind":"REWORK","_number":1,"created":"2023-06-14 08:34:52.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/45/886045/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/45/886045/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/45/886045/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/45/886045/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/45/886045/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/45/886045/1"}}},"commit":{"parents":[{"commit":"1ad65d35d555eec096394f69dbc592c168e56c79","subject":"Merge \"fix: fix the error message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1ad65d35d555eec096394f69dbc592c168e56c79"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-06-14 08:27:38.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-06-14 08:34:36.000000000","tz":480},"subject":"feat: add jwt expire check","message":"feat: add jwt expire check\n\nCheck jwt expire by the time_expired value in the cookie, if current time is bigger than the time_expired value, the request\u0027s header will not has the X-Auth-Token, then the skyline-api will return 401.\n\nChange-Id: I4a33649d3dbf630e36bcdee71462b4796e17f714\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4a6c5132e02848b2ae4ccea84bc493c7a0634eb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4a6c5132e02848b2ae4ccea84bc493c7a0634eb2"}]},"branch":"refs/heads/master"},"2780e41a94e71ad8709e1f1b0052640f93d389ed":{"kind":"REWORK","_number":2,"created":"2023-06-15 01:12:53.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/45/886045/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/45/886045/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/45/886045/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/45/886045/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/45/886045/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/45/886045/2"}}},"commit":{"parents":[{"commit":"4496eb4f05e5bbf5c2e24849c3eb9e4d5e7a9f5a","subject":"feat: update layout styles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4496eb4f05e5bbf5c2e24849c3eb9e4d5e7a9f5a"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-06-14 08:27:38.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-06-15 01:12:35.000000000","tz":480},"subject":"feat: add jwt expire check","message":"feat: add jwt expire check\n\n1. Check jwt expire by the time_expired value in the cookie, if current time is bigger than the time_expired value, the request\u0027s header will not has the X-Auth-Token, then the skyline-api will return 401.\n2. update cypress command to use time_expired cookie\n\nChange-Id: I4a33649d3dbf630e36bcdee71462b4796e17f714\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2780e41a94e71ad8709e1f1b0052640f93d389ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2780e41a94e71ad8709e1f1b0052640f93d389ed"}]},"branch":"refs/heads/master"},"d07227822f26164ba8dee5798a0f675d09995079":{"kind":"REWORK","_number":3,"created":"2023-06-15 02:44:40.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/45/886045/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/45/886045/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/45/886045/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/45/886045/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/45/886045/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/45/886045/3"}}},"commit":{"parents":[{"commit":"4496eb4f05e5bbf5c2e24849c3eb9e4d5e7a9f5a","subject":"feat: update layout styles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4496eb4f05e5bbf5c2e24849c3eb9e4d5e7a9f5a"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-06-14 08:27:38.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-06-15 02:44:27.000000000","tz":480},"subject":"feat: add jwt expire check","message":"feat: add jwt expire check\n\n1. Check jwt expire by the time_expired value in the cookie, if current time is bigger than the time_expired value, the request\u0027s header will not has the X-Auth-Token, then the skyline-api will return 401.\n2. update cypress command to use time_expired cookie\n\nChange-Id: I4a33649d3dbf630e36bcdee71462b4796e17f714\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d07227822f26164ba8dee5798a0f675d09995079"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d07227822f26164ba8dee5798a0f675d09995079"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
