)]}'
{"id":"openstack%2Fskyline-console~886567","triplet_id":"openstack%2Fskyline-console~master~Ie203da67fec57287e4abb1f917782889ade639d3","project":"openstack/skyline-console","branch":"master","topic":"region","attention_set":{},"removed_from_attention_set":{"20865":{"account":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"last_update":"2025-02-11 13:11:59.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2025-02-11 13:11:54.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"Ie203da67fec57287e4abb1f917782889ade639d3","subject":"feat: support multi region","status":"ABANDONED","created":"2023-06-21 04:18:55.000000000","updated":"2026-05-27 03:42:55.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"720db2f11f3c32e3a034bcb02f3ce64417e1bf5a","_number":886567,"virtual_id_number":886567,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-06-05 18:06:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-21 05:54:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-31 11:23:12.000000000","updated_by":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"reviewer":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"state":"CC"},{"updated":"2025-02-11 13:11:23.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-02-11 13:15:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REMOVED"}],"messages":[{"id":"e3d9ee1307f84dbcc98d4b25a4fa650891c3857f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-06-21 04:18:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"38ebed9350d7e86c90faec983179ac727b0c665f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 05:54:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d77c9c067bc6460bb52349307e4518d4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/467c5697944141daa9511e3390f1f5d4 : SUCCESS in 4m 45s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/33afa91ffb204f26b0941edc777e0673 : SUCCESS in 10m 46s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/e40333962fb34f7c9183b6ad14bf5d42 : SUCCESS in 5m 48s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/9914d81b97e4404f856134e709f37ef1 : SUCCESS in 5m 22s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/3f408d1d9fb54c2baebf67375553fa87 : SUCCESS in 1h 25m 12s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/a6a02be868f44477af82d31253851dae : SUCCESS in 46m 15s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/25e8e5aed36143d09f8c89c64b3d94e4 : FAILURE in 44m 15s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/60b64bca507f460aaf5ff11d7e92c3ac : SUCCESS in 1h 02m 28s","accounts_in_message":[],"_revision_number":1},{"id":"53435455f92a159caf0b589dd7b566d98dc8883d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"real_author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2024-06-05 16:37:23.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_30434\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}],"_revision_number":2},{"id":"81f57a5999a1c7f8594275f9a12c749d02799521","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 18:06:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b82777b0e2847d1a21bbf92c8ca328a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46cca3695de44b36aa89917ab561025b : SUCCESS in 3m 50s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/6b71ed2b60bb4141916854980e021380 : SUCCESS in 10m 32s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/5dc0c70f9d384b7fa76d2950d1db8ffc : SUCCESS in 8m 12s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/22c2ced144054d64a29efffdb31499c6 : SUCCESS in 7m 22s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/4b9e9f500d604d6eaef76bc53f8e40f7 : SUCCESS in 1h 23m 02s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/d5cce15bb58d42c7a925174a95c3603a : SUCCESS in 1h 00m 06s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/d9d918f70d8c4c22a469b0b144c6f267 : FAILURE in 23m 51s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/82b795f7f4a94ff8a4fe3e0f4aa463c4 : SUCCESS in 45m 35s","accounts_in_message":[],"_revision_number":2},{"id":"8b84ecd3be664561b4011465fa334a0e7581ca56","author":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"date":"2024-10-31 11:23:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1a456222958d109c68ad8126f16bc4abb5ad19c8","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-11-01 02:15:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7045063271a9b96b4dfaaa4b54bb80ee7775cc2d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-11 13:11:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9c2b57364d0dc3aa18e047a341ed544b920dca1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 13:11:41.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/skyline-console for 886567,2","accounts_in_message":[],"_revision_number":2},{"id":"350bba95672ca0a700435c54d65b575d684b21a5","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-11 13:15:10.000000000","message":"Removed cc \u003cGERRIT_ACCOUNT_37598\u003e.","accounts_in_message":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"_revision_number":2},{"id":"720db2f11f3c32e3a034bcb02f3ce64417e1bf5a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-05-27 03:42:55.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f4a6bc259d321547420ca29d14232b803a888183","revisions":{"3686e56d1236cfc9509f396c4bba31277bb5d232":{"kind":"REWORK","_number":1,"created":"2023-06-21 04:18:55.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/67/886567/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/67/886567/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/67/886567/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/67/886567/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/67/886567/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/67/886567/1"}}},"commit":{"parents":[{"commit":"d07227822f26164ba8dee5798a0f675d09995079","subject":"feat: add jwt expire check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d07227822f26164ba8dee5798a0f675d09995079"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-06-21 04:18:20.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-06-21 04:18:20.000000000","tz":480},"subject":"feat: support multi region","message":"feat: support multi region\n\n1. update domain select to input in the login page\n2. add region name in the header\n3. add region select in the switch project modal\n\nChange-Id: Ie203da67fec57287e4abb1f917782889ade639d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3686e56d1236cfc9509f396c4bba31277bb5d232"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3686e56d1236cfc9509f396c4bba31277bb5d232"}]},"branch":"refs/heads/master"},"f4a6bc259d321547420ca29d14232b803a888183":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-06-05 16:37:23.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"real_uploader":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"ref":"refs/changes/67/886567/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/67/886567/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/67/886567/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/67/886567/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/67/886567/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/67/886567/2"}}},"commit":{"parents":[{"commit":"6c135b55e774ca8350759c9d71cb6093e442e4d7","subject":"Merge \"[translate] add Korean Readme and README Folder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6c135b55e774ca8350759c9d71cb6093e442e4d7"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-06-21 04:18:20.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2024-06-05 16:37:23.000000000","tz":0},"subject":"feat: support multi region","message":"feat: support multi region\n\n1. update domain select to input in the login page\n2. add region name in the header\n3. add region select in the switch project modal\n\nChange-Id: Ie203da67fec57287e4abb1f917782889ade639d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f4a6bc259d321547420ca29d14232b803a888183"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f4a6bc259d321547420ca29d14232b803a888183"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
