)]}'
{"id":"openstack%2Fskyline-console~889371","triplet_id":"openstack%2Fskyline-console~master~I9f0b660fa5457b56f1c5f2d4897e2c4513a3dd5a","project":"openstack/skyline-console","branch":"master","topic":"bugs","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-07-25 08:12:06.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}},"hashtags":[],"change_id":"I9f0b660fa5457b56f1c5f2d4897e2c4513a3dd5a","subject":"fix: update policy for project roles","status":"MERGED","created":"2023-07-25 03:21:19.000000000","updated":"2023-07-25 09:43:53.000000000","submitted":"2023-07-25 09:42:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889371-bugs","meta_rev_id":"03c98fef1d0202a4919e4d9cae1668c244d7d05d","_number":889371,"virtual_id_number":889371,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-25 09:42:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-07-25 05:04:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-07-25 05:04:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-25 04:08:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-25 05:04:01.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"ed75fbcda6e9d3e7377eacb5fc7c659133b1dede","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-07-25 03:21:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8336a5200920d06ad739da2bf3e0ce2e9e60909a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 04:08:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c00a64462cc497588e06d4fa455cfd0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8084fa1d24784923852257f7b4fc9e85 : SUCCESS in 4m 43s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/86d627f63fbc456cb11ab66492308756 : SUCCESS in 7m 55s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/cfa55fa216a240afa2a6b4d295f46b35 : SUCCESS in 7m 26s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/08665961c7b74c3ca3f9cc820369a135 : SUCCESS in 7m 25s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/2574d9e2b69440cda3de7da95c101134 : SUCCESS in 45m 04s","accounts_in_message":[],"_revision_number":1},{"id":"6e2f67b4d231eeea2b4fc2db31c64b0042d72e89","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-07-25 05:04:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"11127865f185df3c882b7525da2afb1afd52be0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 05:04:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dd0ed99180b906db7fede8d49030f875a945df5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 05:48:11.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b5158d0039524db1afe8ffcaa64f31d5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa7775cfd60240d881e42a9a88903327 : SUCCESS in 4m 06s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/bba7d8bb29144d3c9aadbc2c7ee2f27a : SUCCESS in 8m 13s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/6388e6e7ecca4b289cca7a8db36b8a1a : SUCCESS in 6m 59s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/2c46d6ab6baa45468ea65d2035c91f0d : SUCCESS in 5m 08s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/63ecb131fccf42dfaa35fd0b4cce95e4 : FAILURE in 43m 08s","accounts_in_message":[],"_revision_number":1},{"id":"d1894c06b3017d1abacaea7b3d4d538dee1550b0","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-07-25 08:12:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"967c31c42efec102df933debd1a3a69c232b8608","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 08:56:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4794cf0437bb439d94aedffecaa8bef2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c74c9c24bfd4ee29b0d23b019242918 : SUCCESS in 4m 52s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/81767df39b974f0cb0b915944c5f53e5 : SUCCESS in 7m 35s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/154286014f0644829bb2680262e771c3 : SUCCESS in 6m 56s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/ce1fad5d49a544689eff394ce3e80177 : SUCCESS in 5m 20s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/805b4e0206ec443a9433fa2aff9b6233 : SUCCESS in 42m 39s","accounts_in_message":[],"_revision_number":1},{"id":"64211485694b14e0c953f24075759b3f760b4595","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 08:56:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e1ee7aa77a4862bf63c2aa0924eab14a3c998f6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 09:42:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ad14f5c02974451a8d04ff23a37a50f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/019e09e520e844289a383c858d8cb10f : SUCCESS in 10m 52s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/e399fc53ea3a4b1ab5038f071b33ed90 : SUCCESS in 7m 22s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/c30759f3a7214b9ba786c3154fa70bbc : SUCCESS in 5m 19s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/d242390a9ab54a0f9d29e3a440e51716 : SUCCESS in 5m 27s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/58bf59b927ac4ebda057abee8a74c709 : SUCCESS in 45m 07s","accounts_in_message":[],"_revision_number":1},{"id":"3fdfa2fc2d5a8d276bef5ff5932ec63aa3511bfa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 09:42:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"03c98fef1d0202a4919e4d9cae1668c244d7d05d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 09:43:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71bca8dff622486caed777a3c22b2212\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/246a4306074b486f820863b41174d0c0 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1f5940fd1164c6d6435b6470221c1ee0f8a77704","revisions":{"1f5940fd1164c6d6435b6470221c1ee0f8a77704":{"kind":"REWORK","_number":1,"created":"2023-07-25 03:21:19.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/71/889371/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/71/889371/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/71/889371/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/71/889371/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/71/889371/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/71/889371/1"}}},"commit":{"parents":[{"commit":"206ffc9e9266d8fd1a50219d89266b3fac5521f4","subject":"fix: fix the hints of allocation pool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/206ffc9e9266d8fd1a50219d89266b3fac5521f4"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-07-25 03:19:48.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-07-25 03:19:48.000000000","tz":480},"subject":"fix: update policy for project roles","message":"fix: update policy for project roles\n\n1. update policy for project manage users\n2. update policy for project manage user groups\n\nChange-Id: I9f0b660fa5457b56f1c5f2d4897e2c4513a3dd5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1f5940fd1164c6d6435b6470221c1ee0f8a77704"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1f5940fd1164c6d6435b6470221c1ee0f8a77704"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
