)]}'
{"id":"openstack%2Fskyline-console~890399","triplet_id":"openstack%2Fskyline-console~master~I6ece6d4cf50fbed3ac2569b70fc47591932d69c6","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"36246":{"account":{"_account_id":36246,"name":"liyoumin","display_name":"liyoumin","email":"306917629@qq.com","username":"liym"},"last_update":"2023-08-04 06:12:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6ece6d4cf50fbed3ac2569b70fc47591932d69c6","subject":"fix: fix the domain description","status":"MERGED","created":"2023-08-03 09:14:09.000000000","updated":"2023-08-04 06:13:33.000000000","submitted":"2023-08-04 06:12:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890399","meta_rev_id":"bcf252cb0986e705fe0acba3d9aabaccc517c77f","_number":890399,"virtual_id_number":890399,"owner":{"_account_id":36246,"name":"liyoumin","display_name":"liyoumin","email":"306917629@qq.com","username":"liym"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-04 06:12:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2023-08-04 05:22:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2023-08-04 05:22:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-03 10:27:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-04 05:22:15.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"bfc6748b3260e083dc1a56affad06aa442be21a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36246,"name":"liyoumin","display_name":"liyoumin","email":"306917629@qq.com","username":"liym"},"date":"2023-08-03 09:14:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d36002e16431e6ef01b4db4056d3e125c6e6f780","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 10:27:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed921dc31f394070940d62fe9929b5db\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5eb423b596b4568894e44a41147a10f : SUCCESS in 4m 55s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/86c0537993f24dbbb9f36740c8cd3b42 : SUCCESS in 11m 00s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/259bd8085a194e4985667a569b18a725 : SUCCESS in 6m 46s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/7d97baadb4494e05bfe06d1969bdf71f : SUCCESS in 5m 32s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/8bd52753dca64ac4a63b20b8a77bcf6a : SUCCESS in 1h 05m 34s","accounts_in_message":[],"_revision_number":1},{"id":"25d61041431f93d422505357d0fb18711dc740fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36246,"name":"liyoumin","display_name":"liyoumin","email":"306917629@qq.com","username":"liym"},"date":"2023-08-04 01:16:23.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"8a5664775fab981a8e27b81d6afe3aaa635c6e9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36246,"name":"liyoumin","display_name":"liyoumin","email":"306917629@qq.com","username":"liym"},"date":"2023-08-04 01:22:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7b58f4462b6c6e68ceed00fd1c1142987d88608b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36246,"name":"liyoumin","display_name":"liyoumin","email":"306917629@qq.com","username":"liym"},"date":"2023-08-04 01:23:46.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a2abad0c70b546e6c7c07f87f946b59583d33e88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 02:26:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7be88d707ec846c48c137f7ec81e72eb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/073ec31124f24da4a2e8833d6c26cbdc : SUCCESS in 4m 11s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/46513e095ec6492089d0a8b9d3ebe4ab : SUCCESS in 7m 21s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/dc69851988514c8d8379cc0e2414d332 : SUCCESS in 8m 09s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/f32184bac46d460e8b0710f256399547 : SUCCESS in 7m 30s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/faea12178fea45a1a7a695c2cdd83be9 : SUCCESS in 57m 10s","accounts_in_message":[],"_revision_number":4},{"id":"3c9b61a95ef9e5aebd5a648fb38e68a94ca763b0","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-08-04 05:22:15.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"1fc6d481de2ef6ea38efc34d2c8114d40e714062","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 05:22:44.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ed3211b9e96d51105a5c63d7eebf7084b5ca25d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 06:12:31.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8165a1bc0bc4679a33f16cc1717bfa4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13addb4bd2994182ad15726aa6cfda54 : SUCCESS in 5m 07s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/c7b72c555b274b63b8598457789d3055 : SUCCESS in 8m 07s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/c8a42d462d39484ebfc5bdabcf955468 : SUCCESS in 7m 54s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/f1697799ba5548f08199ec422d840b1c : SUCCESS in 7m 36s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/603e60cb93374bdfae0574f9d6753e1b : SUCCESS in 48m 44s","accounts_in_message":[],"_revision_number":4},{"id":"2a4f947307478624ac731957f0e2a34162a13c32","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 06:12:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"bcf252cb0986e705fe0acba3d9aabaccc517c77f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 06:13:33.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/faec91a9fc5844eab0d1cbb7fbfa8f04\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d1308aafee34fe4bf12de01c6ccc193 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"cfe389dd3981252c7c6350a6237ce8515e193e3c","revisions":{"ed24033c3f2ef772bdee3750db99e29f0998ab0d":{"kind":"REWORK","_number":1,"created":"2023-08-03 09:14:09.000000000","uploader":{"_account_id":36246,"name":"liyoumin","display_name":"liyoumin","email":"306917629@qq.com","username":"liym"},"ref":"refs/changes/99/890399/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/99/890399/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/99/890399/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/99/890399/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/99/890399/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/99/890399/1"}}},"commit":{"parents":[{"commit":"a95c588a5d84427b7f025f661c974a197d669300","subject":"fix: fix table header buttons style","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a95c588a5d84427b7f025f661c974a197d669300"}]}],"author":{"name":"liyoumin","email":"306917629@qq.com","date":"2023-08-03 09:13:50.000000000","tz":480},"committer":{"name":"liyoumin","email":"306917629@qq.com","date":"2023-08-03 09:13:50.000000000","tz":480},"subject":"fix: fix the domain description","message":"fix: fix the domain description\n\n1. fix the domain description\n2.filter the disabled domain when get domain options\n\nChange-Id: I6ece6d4cf50fbed3ac2569b70fc47591932d69c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ed24033c3f2ef772bdee3750db99e29f0998ab0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ed24033c3f2ef772bdee3750db99e29f0998ab0d"}]},"branch":"refs/heads/master"},"c1c173bc98a422d1c344c4fce40d73ae3883c6c4":{"kind":"REWORK","_number":2,"created":"2023-08-04 01:16:23.000000000","uploader":{"_account_id":36246,"name":"liyoumin","display_name":"liyoumin","email":"306917629@qq.com","username":"liym"},"ref":"refs/changes/99/890399/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/99/890399/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/99/890399/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/99/890399/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/99/890399/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/99/890399/2"}}},"commit":{"parents":[{"commit":"a95c588a5d84427b7f025f661c974a197d669300","subject":"fix: fix table header buttons style","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a95c588a5d84427b7f025f661c974a197d669300"}]}],"author":{"name":"liyoumin","email":"306917629@qq.com","date":"2023-08-03 09:13:50.000000000","tz":480},"committer":{"name":"liyoumin","email":"306917629@qq.com","date":"2023-08-04 01:16:02.000000000","tz":480},"subject":"fix: fix the domain description","message":"fix: fix the domain description\n\n1. fix the domain description\n2.filter the disabled domain when get domain options\n\nChange-Id: I6ece6d4cf50fbed3ac2569b70fc47591932d69c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c1c173bc98a422d1c344c4fce40d73ae3883c6c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c1c173bc98a422d1c344c4fce40d73ae3883c6c4"}]},"branch":"refs/heads/master"},"68cafaef5815fcb4208c7f8177cbec633f6eacb2":{"kind":"REWORK","_number":3,"created":"2023-08-04 01:22:20.000000000","uploader":{"_account_id":36246,"name":"liyoumin","display_name":"liyoumin","email":"306917629@qq.com","username":"liym"},"ref":"refs/changes/99/890399/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/99/890399/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/99/890399/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/99/890399/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/99/890399/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/99/890399/3"}}},"commit":{"parents":[{"commit":"a95c588a5d84427b7f025f661c974a197d669300","subject":"fix: fix table header buttons style","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a95c588a5d84427b7f025f661c974a197d669300"}]}],"author":{"name":"liyoumin","email":"306917629@qq.com","date":"2023-08-03 09:13:50.000000000","tz":480},"committer":{"name":"liyoumin","email":"306917629@qq.com","date":"2023-08-04 01:22:01.000000000","tz":480},"subject":"fix: fix the domain description","message":"fix: fix the domain description\n\n1. fix the domain description\n2.filter the disabled domain when get domain options\n\nChange-Id: I6ece6d4cf50fbed3ac2569b70fc47591932d69c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/68cafaef5815fcb4208c7f8177cbec633f6eacb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/68cafaef5815fcb4208c7f8177cbec633f6eacb2"}]},"branch":"refs/heads/master"},"cfe389dd3981252c7c6350a6237ce8515e193e3c":{"kind":"REWORK","_number":4,"created":"2023-08-04 01:23:46.000000000","uploader":{"_account_id":36246,"name":"liyoumin","display_name":"liyoumin","email":"306917629@qq.com","username":"liym"},"ref":"refs/changes/99/890399/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/99/890399/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/99/890399/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/99/890399/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/99/890399/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/99/890399/4"}}},"commit":{"parents":[{"commit":"a95c588a5d84427b7f025f661c974a197d669300","subject":"fix: fix table header buttons style","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a95c588a5d84427b7f025f661c974a197d669300"}]}],"author":{"name":"liyoumin","email":"306917629@qq.com","date":"2023-08-03 09:13:50.000000000","tz":480},"committer":{"name":"liyoumin","email":"306917629@qq.com","date":"2023-08-04 01:23:21.000000000","tz":480},"subject":"fix: fix the domain description","message":"fix: fix the domain description\n\n1. fix the domain description\n2.filter the disabled domain when get domain options\n\nChange-Id: I6ece6d4cf50fbed3ac2569b70fc47591932d69c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cfe389dd3981252c7c6350a6237ce8515e193e3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cfe389dd3981252c7c6350a6237ce8515e193e3c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
