)]}'
{"id":"openstack%2Fskyline-console~890628","triplet_id":"openstack%2Fskyline-console~master~I2b50699427c836e2dfb098c9f42c7ae06b385a34","project":"openstack/skyline-console","branch":"master","topic":"890628","attention_set":{},"removed_from_attention_set":{"20865":{"account":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"last_update":"2023-12-04 09:05:46.000000000","reason":"Change was submitted"},"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2023-12-04 09:05:46.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-12-04 07:31:24.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"34680":{"account":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"last_update":"2023-12-04 09:05:46.000000000","reason":"Change was submitted"},"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2023-12-04 09:05:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2b50699427c836e2dfb098c9f42c7ae06b385a34","subject":"Adding Turkish translation","status":"MERGED","created":"2023-08-07 07:26:35.000000000","updated":"2023-12-04 09:06:45.000000000","submitted":"2023-12-04 09:05:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"890628-890628","meta_rev_id":"45137f482d5bcef5f5ef74e17268bf24c56d42f1","_number":890628,"virtual_id_number":890628,"owner":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-04 09:05:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":0,"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"value":2,"date":"2023-12-04 07:31:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":0,"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"value":1,"date":"2023-12-04 07:31:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"}],"reviewers":{"REVIEWER":[{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-07 10:06:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-07 14:48:50.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2023-09-13 12:44:05.000000000","updated_by":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"reviewer":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"state":"REVIEWER"},{"updated":"2023-09-13 12:44:05.000000000","updated_by":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"},{"updated":"2023-11-28 08:25:06.000000000","updated_by":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"reviewer":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"state":"REVIEWER"}],"messages":[{"id":"574b8904651661007eeaf15e77ea5bd574d69f77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"date":"2023-08-07 07:26:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"84e634e34741f5f48c88e04034225d04ad14b773","author":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"date":"2023-08-07 07:39:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"640b6e336b02ac32235742f65a0dfab4249f5201","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 10:06:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f45226d12a643f7a4057d916fe60d3f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b29ec9e2059421387e1452ee6d904d8 : SUCCESS in 4m 51s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/70ab5cd363a94d568561221cd875a6ab : FAILURE in 7m 35s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/7e33c7b666654e2aa0661b476133eeb6 : SUCCESS in 5m 44s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/b290a0ca5c834d8e939147305b0752f6 : FAILURE in 5m 20s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/20972bf3bf4a4867984a2fa4637127ec : FAILURE in 2h 11m 14s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/c2dcf14f853c44d299171f978fd98178 : FAILURE in 1h 21m 21s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/495f948e64b043d6a4d79a10e9c84004 : FAILURE in 2h 32m 06s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/7d6e5e4f7d4f47fa9813bd6f095afab1 : FAILURE in 1h 37m 21s","accounts_in_message":[],"_revision_number":1},{"id":"d59cea9298f4e1d99d1edac2d31716622b75c107","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"date":"2023-08-07 12:56:36.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"adb8494db30efea4ea94c203ee3f94b15c0759bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 14:16:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/521d808da7a9447785f251c209d952bf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a65aebe1cb3c4bf886a09cad9b3e1436 : SUCCESS in 4m 43s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/beac56f1d583414da6a62c446e8d1b75 : SUCCESS in 10m 17s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/88366ea984484fb983e4f5a9f0480bc9 : SUCCESS in 8m 03s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/848c4a9c01ab4c0d97dc96bfef803792 : SUCCESS in 7m 31s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/5439df2433d2447a9e617b9ff1aea6a1 : SUCCESS in 1h 04m 11s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/83b54691affa4060a557c835c8cb46cf : SUCCESS in 1h 09m 00s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/f9efb888b5e84c8793ced6981e603d2d : SUCCESS in 1h 04m 02s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/8a35d7a6bf2a4fbaa39bcdad0d3e7066 : SUCCESS in 1h 10m 33s","accounts_in_message":[],"_revision_number":2},{"id":"398fab5a21c651e186001de7a210183d0b0a5a18","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-08-07 14:48:50.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e6a0a43733189e7a6f9f203723aec623e9fe762a","author":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"date":"2023-08-08 08:26:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1f5eb641f30642d44607245fb422d8b684d2ee2f","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-08-11 08:18:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0163a1c6ae0d8736902a98a49866c6496cd4405b","author":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"date":"2023-08-14 06:37:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2ce5e6fd4cb16584a26cc218a710f03e75b72733","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-08-18 01:40:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f9345c669f5ba7ca97cab062d6c526ecc3fc43e9","author":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"date":"2023-09-04 08:04:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"664f6aa95bd70a1fa799e516320b853b765c6ff1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"date":"2023-09-13 12:18:07.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"3be9b125b24276c05f5671858169d8232acee20c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 12:18:24.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/skyline-console for 890628,3","accounts_in_message":[],"_revision_number":3},{"id":"d354d9eb7ebd1de15be6924c43787f6a70feaebc","author":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"date":"2023-09-13 12:20:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6f8b6979d14769be923d8cbe1d23cbaa20fdb8ad","author":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"date":"2023-09-13 12:46:26.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d5a4e817d0fc20172f08447ff8a12730d74ac4e0","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-09-15 02:18:10.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4fa018967e9055c9b0a4ab63324a9268bc228554","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"date":"2023-11-28 06:09:31.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"e5deaf5b06a4be55ebfa8d1199c0d4bf091105cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 07:56:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/055acbf42e4e47258e89a0652ef33de5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/323c7a919d8141bfa2355a6ce5550d63 : SUCCESS in 3m 51s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/f25e355e840249aabc2ee8eb93a15c87 : SUCCESS in 7m 43s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/364969885b2549d3955f988575e49ff3 : SUCCESS in 7m 26s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/8e7ef089ada84c7db1d9608276609a17 : SUCCESS in 5m 15s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/4b9f558458ec4cf3809523ccf4da69c4 : SUCCESS in 1h 39m 02s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/7ae278de232e4338bf035df23b9af0f9 : SUCCESS in 58m 46s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/3611a02f43f346cea9d28ac8e38c1a96 : SUCCESS in 48m 03s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/732ba24ac4a6437aa1090493e26641ab : SUCCESS in 1h 03m 18s","accounts_in_message":[],"_revision_number":4},{"id":"ce17e7d34df69298804b9d2b206e5fded975934d","author":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"date":"2023-11-28 08:25:06.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"0cd823d5b8ca6420c90c3af4986be43c9a62a15a","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-11-29 00:59:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fb1f64bc2500a6347e5652d7711ae9d4fa669336","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"date":"2023-12-04 06:02:24.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"1239aa50805682bfbd1cfeeae4d6ccb38c5882e8","author":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"date":"2023-12-04 06:04:58.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bfbf130fd90b1df6aeebf86889d31788265c2d8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 07:16:14.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2c3d35d2e0b4be48a6b55c53b9b9698\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62169a0b34b34d16b7476ed23080dc9a : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f95a96f91cc94aeaa5654604e54a7c3f : SUCCESS in 6m 53s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/a1e9ff1fdaa64f94964e93599bf1ea39 : SUCCESS in 6m 48s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/c5932ac2463744aea128dcfc470a3c56 : SUCCESS in 6m 45s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/a0adb1fb0a464703a147b27e17ebf279 : SUCCESS in 7m 18s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/99d2cd6989c447c4b8884acf0842a43d : SUCCESS in 1h 11m 00s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/73e9e4ca78b44847805da16b0bf69b26 : SUCCESS in 53m 09s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/3f1dbb1e3aa44327948eca0fe306a51e : SUCCESS in 59m 40s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/d02a6066ce5d445b9c90a48634703a4c : SUCCESS in 53m 18s","accounts_in_message":[],"_revision_number":5},{"id":"bc64bdd3f08b751920de6a4acde3231e45cf3d36","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-12-04 07:31:24.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"68e8f005cec82cf21699e3a59d06fa3a10b6445c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 07:32:03.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"8c3ce2512a38906bb311f669cf8de2db9ea84864","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 09:05:46.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e22bb280205f4f0683eec8a0210cd548\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/776e6d83ed8f447ea0552a26fb0423de : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd63571493b445e2b9e741bf1ffc1daf : SUCCESS in 4m 32s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/ea2e70ad47fb4831a4410ad7180f36c7 : SUCCESS in 7m 36s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/9fb0a24b6295466ca2cefc488632d539 : SUCCESS in 7m 29s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/f613169cbcd7442083590b2564d5bf22 : SUCCESS in 5m 34s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/a711cdd72ffb4e9b959a09e7fda26d6c : SUCCESS in 1h 28m 21s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/61c9a51152b741a6af4f2f4d7b357876 : SUCCESS in 50m 00s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/52862460db4c4e7e8968ebc0e7b71926 : SUCCESS in 1h 02m 22s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/c98a7677b42f465380a1aaf973f40820 : SUCCESS in 1h 01m 02s","accounts_in_message":[],"_revision_number":5},{"id":"3cf1656726aec8672678ff9a008fe989637944ef","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 09:05:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"45137f482d5bcef5f5ef74e17268bf24c56d42f1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 09:06:45.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c79bed5d96354eaeaad4f9e930b672f1\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c6853d48b3349018b13fb488736fca1 : SUCCESS in 41s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8502e2f2fca40ec967094cb53fd463e : SUCCESS in 41s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"4d606d4742df5a81afb22d57ce7eb152ca126abd","revisions":{"9232121c99846622b02db9834e0692e1ca7dcc2d":{"kind":"REWORK","_number":1,"created":"2023-08-07 07:26:35.000000000","uploader":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"ref":"refs/changes/28/890628/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/28/890628/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/890628/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/890628/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/890628/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/28/890628/1"}}},"commit":{"parents":[{"commit":"cfe389dd3981252c7c6350a6237ce8515e193e3c","subject":"fix: fix the domain description","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cfe389dd3981252c7c6350a6237ce8515e193e3c"}]}],"author":{"name":"Numan MENEKŞE","email":"numan.menekse@tubitak.gov.tr","date":"2023-08-07 07:22:28.000000000","tz":0},"committer":{"name":"Numan MENEKŞE","email":"numan.menekse@tubitak.gov.tr","date":"2023-08-07 07:25:36.000000000","tz":0},"subject":"Adding Turkish translation","message":"Adding Turkish translation\n\nChange-Id: I2b50699427c836e2dfb098c9f42c7ae06b385a34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9232121c99846622b02db9834e0692e1ca7dcc2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9232121c99846622b02db9834e0692e1ca7dcc2d"}]},"branch":"refs/heads/master"},"a0ba1d764b7cd5ec414c6d0145bd12c3fb53deb0":{"kind":"REWORK","_number":2,"created":"2023-08-07 12:56:36.000000000","uploader":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"ref":"refs/changes/28/890628/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/28/890628/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/890628/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/890628/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/890628/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/28/890628/2"}}},"commit":{"parents":[{"commit":"cfe389dd3981252c7c6350a6237ce8515e193e3c","subject":"fix: fix the domain description","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cfe389dd3981252c7c6350a6237ce8515e193e3c"}]}],"author":{"name":"Numan MENEKŞE","email":"numan.menekse@tubitak.gov.tr","date":"2023-08-07 07:22:28.000000000","tz":0},"committer":{"name":"Numan MENEKŞE","email":"numan.menekse@tubitak.gov.tr","date":"2023-08-07 12:55:30.000000000","tz":0},"subject":"Adding Turkish translation","message":"Adding Turkish translation\n\nChange-Id: I2b50699427c836e2dfb098c9f42c7ae06b385a34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a0ba1d764b7cd5ec414c6d0145bd12c3fb53deb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a0ba1d764b7cd5ec414c6d0145bd12c3fb53deb0"}]},"branch":"refs/heads/master"},"0d75c8f1edd31969e714542ea9ef2fe191a856e4":{"kind":"REWORK","_number":3,"created":"2023-09-13 12:18:07.000000000","uploader":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"ref":"refs/changes/28/890628/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/28/890628/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/890628/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/890628/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/890628/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/28/890628/3"}}},"commit":{"parents":[{"commit":"cfe389dd3981252c7c6350a6237ce8515e193e3c","subject":"fix: fix the domain description","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cfe389dd3981252c7c6350a6237ce8515e193e3c"}]}],"author":{"name":"Numan MENEKŞE","email":"numan.menekse@tubitak.gov.tr","date":"2023-08-07 07:22:28.000000000","tz":0},"committer":{"name":"Numan MENEKŞE","email":"numan.menekse@tubitak.gov.tr","date":"2023-09-13 12:16:58.000000000","tz":0},"subject":"Adding Turkish translation","message":"Adding Turkish translation\n\nChange-Id: I2b50699427c836e2dfb098c9f42c7ae06b385a34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0d75c8f1edd31969e714542ea9ef2fe191a856e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0d75c8f1edd31969e714542ea9ef2fe191a856e4"}]},"branch":"refs/heads/master"},"e2a2a3ad84cdc498de17363968cd8d4c610e2217":{"kind":"REWORK","_number":4,"created":"2023-11-28 06:09:31.000000000","uploader":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"ref":"refs/changes/28/890628/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/28/890628/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/890628/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/890628/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/890628/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/28/890628/4"}}},"commit":{"parents":[{"commit":"5a3156331681f2468c5572f96c5971ba0d8f4128","subject":"fix: fix force shutdown form item in the resize instance form","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5a3156331681f2468c5572f96c5971ba0d8f4128"}]}],"author":{"name":"Numan MENEKŞE","email":"numan.menekse@tubitak.gov.tr","date":"2023-08-07 07:22:28.000000000","tz":0},"committer":{"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","date":"2023-11-28 06:08:43.000000000","tz":0},"subject":"Adding Turkish translation","message":"Adding Turkish translation\n\nChange-Id: I2b50699427c836e2dfb098c9f42c7ae06b385a34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e2a2a3ad84cdc498de17363968cd8d4c610e2217"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e2a2a3ad84cdc498de17363968cd8d4c610e2217"}]},"branch":"refs/heads/master"},"4d606d4742df5a81afb22d57ce7eb152ca126abd":{"kind":"REWORK","_number":5,"created":"2023-12-04 06:02:24.000000000","uploader":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"ref":"refs/changes/28/890628/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/28/890628/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/890628/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/890628/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/890628/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/28/890628/5"}}},"commit":{"parents":[{"commit":"5a3156331681f2468c5572f96c5971ba0d8f4128","subject":"fix: fix force shutdown form item in the resize instance form","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5a3156331681f2468c5572f96c5971ba0d8f4128"}]}],"author":{"name":"Numan MENEKŞE","email":"numan.menekse@tubitak.gov.tr","date":"2023-08-07 07:22:28.000000000","tz":0},"committer":{"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","date":"2023-12-04 05:58:06.000000000","tz":0},"subject":"Adding Turkish translation","message":"Adding Turkish translation\n\nChange-Id: I2b50699427c836e2dfb098c9f42c7ae06b385a34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4d606d4742df5a81afb22d57ce7eb152ca126abd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4d606d4742df5a81afb22d57ce7eb152ca126abd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
