)]}'
{"id":"openstack%2Fskyline-console~890649","triplet_id":"openstack%2Fskyline-console~master~I9aa223951e610e9a63e5fdc680381aea88d08ff0","project":"openstack/skyline-console","branch":"master","topic":"instance","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-08-07 16:53:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9aa223951e610e9a63e5fdc680381aea88d08ff0","subject":"refactor: refactor create ironic","status":"MERGED","created":"2023-08-07 12:28:25.000000000","updated":"2023-08-07 16:55:03.000000000","submitted":"2023-08-07 16:53:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890649-instance","meta_rev_id":"9a0c8eb600049ae81b763e670866d0ddb9618ce3","_number":890649,"virtual_id_number":890649,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-07 16:53:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-08-07 15:16:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-08-07 15:16:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-07 13:46:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-07 15:16:42.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"f3f3b66f3da84a8dd5a5fc1b78816d24aa389e6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-08-07 12:28:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ecc238b87e7fb806f038c3ab330f5ef0e9aa4c8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 13:46:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e8a9723dcb74aaaa6d0d548bcefce32\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09e376f1b23c4ae9bff9bb2ffa03c420 : SUCCESS in 4m 42s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/1743d528f5c8469e8f622988bc18cd6a : SUCCESS in 10m 27s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/a2f05abf6338406d9614df025817be8e : SUCCESS in 6m 55s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/6e2d22b6ea654fd1a2f5b2abe06d7f5d : SUCCESS in 7m 53s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/8e51c5c4ea7c4c7a91981a81ad05e859 : SUCCESS in 1h 11m 40s","accounts_in_message":[],"_revision_number":1},{"id":"9e3b5936da4ed78db59dd5d4d9db51e631e3f320","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-08-07 15:16:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2695f299ea573581428e3e33c999c4851fddca6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 15:17:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"913e2be8a40d8e71a271ca27d5f3c3f470b39819","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 16:53:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90f6ef96b5ad4bdea4965012ff1e6abd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e29fc95031714fcea108741b30d9ff44 : SUCCESS in 4m 26s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/488f5ab3301f46c1ad36a8fc53ef2e78 : SUCCESS in 12m 47s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/ee531d6f10ee461aad13aff627cbe364 : SUCCESS in 8m 04s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/a57c5079c7de402aa2edbb276136b651 : SUCCESS in 6m 22s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/298ad15564674d058c1474aa4ec41679 : SUCCESS in 1h 30m 19s","accounts_in_message":[],"_revision_number":1},{"id":"0031aabcc89188dbb70d830c4a7a1b4dc80396dd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 16:53:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9a0c8eb600049ae81b763e670866d0ddb9618ce3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 16:55:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c22b9640f2c4e01bdaa79abe09c96f6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fee452e3263482c80b49973e6702814 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"06e581191b2878b3feaf38cb3f196d02684cbecd","revisions":{"06e581191b2878b3feaf38cb3f196d02684cbecd":{"kind":"REWORK","_number":1,"created":"2023-08-07 12:28:25.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/49/890649/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/49/890649/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/49/890649/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/49/890649/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/49/890649/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/49/890649/1"}}},"commit":{"parents":[{"commit":"29c4030d0c178947c9f08e194866497e84c6fd6b","subject":"refactor: refactor create instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/29c4030d0c178947c9f08e194866497e84c6fd6b"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-08-07 12:27:21.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-08-07 12:27:28.000000000","tz":480},"subject":"refactor: refactor create ironic","message":"refactor: refactor create ironic\n\nrefact create ironic code\n\nChange-Id: I9aa223951e610e9a63e5fdc680381aea88d08ff0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/06e581191b2878b3feaf38cb3f196d02684cbecd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/06e581191b2878b3feaf38cb3f196d02684cbecd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
