)]}'
{"id":"openstack%2Fskyline-console~891666","triplet_id":"openstack%2Fskyline-console~master~Id7492e49668fcd4c6408363ad636e45ba8a67ae8","project":"openstack/skyline-console","branch":"master","topic":"delete","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-08-19 00:52:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id7492e49668fcd4c6408363ad636e45ba8a67ae8","subject":"fix: update delete tip for instance soft delete","status":"MERGED","created":"2023-08-17 09:27:28.000000000","updated":"2023-08-19 00:53:13.000000000","submitted":"2023-08-19 00:52:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"891666-delete","meta_rev_id":"b24defd4c966477070e8a0994a27279bc492ee24","_number":891666,"virtual_id_number":891666,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-19 00:52:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-08-18 23:46:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-08-18 23:46:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-17 10:50:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-18 07:42:18.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"CC"},{"updated":"2023-08-18 23:46:55.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"7f72bca888bca0748e33a22fbebd1ee8decd4286","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-08-17 09:27:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"205e89220bdbc3b8f7954a04f19233a9ec6ee907","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-08-17 09:34:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a2c0c56e0671cc8733c34154c79d1155b28c0143","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 10:50:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c9e175cc74948f1910729d245b7eb0d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b645e7c005b4e46a611155fb2c25d08 : SUCCESS in 6m 30s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/538ac3916ba94f2caa4691f8711abbaf : SUCCESS in 7m 21s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/85b2732421d3480b877c8374e6bf22ae : SUCCESS in 7m 01s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/48b1d714927f4b0d95ce2fd4e8234946 : SUCCESS in 7m 23s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/1a0be6a50093445893be076fa335b527 : FAILURE in 1h 04m 03s","accounts_in_message":[],"_revision_number":2},{"id":"38a9eb32a08bf4725900278c4a9e91ef4a8a7996","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-08-18 07:42:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b39574e7ac864188294cd22981c70e8375cf88af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 09:18:28.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4901574dac23490499c24cfad4350207\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6e5e08bd88c484ca4995358a12c3f3a : SUCCESS in 4m 33s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/f805afb8286a4785ae38445b66bd8600 : SUCCESS in 11m 00s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/fb59fc5f95e44471b293d6ba3d8fd128 : SUCCESS in 4m 42s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/d1c0372671df4c4cbdb24c5f1850f30d : SUCCESS in 7m 07s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/db8e48ce29574e3bad46bac62ff85e56 : FAILURE in 1h 30m 45s","accounts_in_message":[],"_revision_number":2},{"id":"905deb0588fa6de79693d7ee915b9c2e1b293ea9","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-08-18 09:24:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"855c5319fbfef26c58043958cd7bc8d520192472","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 11:04:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ee0057c3a9e46fa8d2b1f81dd15a29b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7034fc3df184e31bad307159906ad98 : SUCCESS in 4m 12s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/c95b72ead03b4b87b226a8452a867bd2 : SUCCESS in 7m 06s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/809306ce57134d3197b6671c26c51914 : SUCCESS in 5m 27s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/f0b7c6a7cbe44c9daecf6a8b21a6f7a3 : SUCCESS in 7m 52s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/fe669235bccf4832b0f44300fb3b2714 : SUCCESS in 1h 33m 56s","accounts_in_message":[],"_revision_number":2},{"id":"77e1c335cbc9c7aa8cde22cfc5a68e0bed500a5c","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-08-18 23:46:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"eee49d8d99ec1ae954367c417591a2860db244ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 23:47:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"11971722ba46845f03476f6e539bd553359c433e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-19 00:52:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14393d14fe1f4fda8e92c0eb2bc8ff84\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a9eb8e61b7444f3ab83764241c19ebc : SUCCESS in 4m 30s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/a408b74b644542dd9f9a2f3920f1473d : SUCCESS in 10m 00s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/efca895cc3c245e3919eb08046a31278 : SUCCESS in 7m 21s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/e57e701e63074c469d83c603d6cdf22c : SUCCESS in 8m 10s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/9ac6e5faba004fa8bb473b1218ed4bea : SUCCESS in 1h 03m 01s","accounts_in_message":[],"_revision_number":2},{"id":"c129c04efaed1535d24fcbcce45613ebf7fbd0b8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-19 00:52:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b24defd4c966477070e8a0994a27279bc492ee24","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-19 00:53:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e77eba500c4042159359de37f7af35aa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd22072594db4cc38f1d338a9f507717 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3f23280df53142d83eda4ac9098d4420824deac4","revisions":{"6514f6863df552b2b39e262b91f7933ed4fab499":{"kind":"REWORK","_number":1,"created":"2023-08-17 09:27:28.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/66/891666/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/66/891666/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/66/891666/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/66/891666/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/66/891666/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/66/891666/1"}}},"commit":{"parents":[{"commit":"b36d5593e6d68e02c6ba0e88e54d1fa0f2b49bb5","subject":"feat: support custom locale language","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b36d5593e6d68e02c6ba0e88e54d1fa0f2b49bb5"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-08-17 09:27:03.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-08-17 09:27:03.000000000","tz":480},"subject":"fix: update delete tip for instance soft delete","message":"fix: update delete tip for instance soft delete\n\nupdate delete tip for soft delete instance\n\nChange-Id: Id7492e49668fcd4c6408363ad636e45ba8a67ae8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6514f6863df552b2b39e262b91f7933ed4fab499"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6514f6863df552b2b39e262b91f7933ed4fab499"}]},"branch":"refs/heads/master"},"3f23280df53142d83eda4ac9098d4420824deac4":{"kind":"REWORK","_number":2,"created":"2023-08-17 09:34:21.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/66/891666/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/66/891666/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/66/891666/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/66/891666/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/66/891666/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/66/891666/2"}}},"commit":{"parents":[{"commit":"b36d5593e6d68e02c6ba0e88e54d1fa0f2b49bb5","subject":"feat: support custom locale language","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b36d5593e6d68e02c6ba0e88e54d1fa0f2b49bb5"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-08-17 09:27:03.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-08-17 09:33:53.000000000","tz":480},"subject":"fix: update delete tip for instance soft delete","message":"fix: update delete tip for instance soft delete\n\nupdate delete tip for soft delete instance\n\nChange-Id: Id7492e49668fcd4c6408363ad636e45ba8a67ae8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3f23280df53142d83eda4ac9098d4420824deac4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3f23280df53142d83eda4ac9098d4420824deac4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
