)]}'
{"id":"openstack%2Fskyline-console~892482","triplet_id":"openstack%2Fskyline-console~master~I50601a6b2bd9b36b0835fdb104f4d66b1707e728","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-08-23 10:17:22.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2023-08-23 11:34:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I50601a6b2bd9b36b0835fdb104f4d66b1707e728","subject":"fix: fix the magic input","status":"MERGED","created":"2023-08-23 08:07:45.000000000","updated":"2023-08-23 11:35:34.000000000","submitted":"2023-08-23 11:34:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892482","meta_rev_id":"8659f7bc3fec9997b537d1b62d31ff7c2862c467","_number":892482,"virtual_id_number":892482,"owner":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-23 11:34:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2023-08-23 10:17:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2023-08-23 10:17:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-23 10:10:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-23 10:13:32.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"89159cdd9664f1e4f115cf905fef144147026dc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-08-23 08:07:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a4a716366dd1807d1f9636bb43a9996456c3f1a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 10:10:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f591edbd77a4570a61aa0676768baf1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28895c6cb6ce44cbb4624fded8899958 : SUCCESS in 4m 08s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/ad5216fcba0e4264bfd08c122b026998 : SUCCESS in 10m 11s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/2ad8fc1f25c8470bb212a922585d4397 : SUCCESS in 7m 46s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/156125f01a004c459e69eb467a561492 : SUCCESS in 7m 39s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/6db8ffd107bc4627a460e3390071fbf2 : SUCCESS in 1h 30m 38s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/ba49c5a30a5a44d5980a96cafb02fa84 : SUCCESS in 56m 22s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/8eee88482968421aaa72f4bf68e505ba : SUCCESS in 57m 05s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/b3d3967064754ea2a31e764cc98e5c69 : SUCCESS in 43m 09s","accounts_in_message":[],"_revision_number":1},{"id":"6405802df5a45bc519b84ae98ab0ea2ca7d3918f","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-08-23 10:17:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d791ac813a6468ae62cc1968de7c17bfc24f03ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 10:17:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5ee47582eebb2676954ced62eed5a1e58485a7a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 11:34:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b6512f81d3c4bcd9386902a0e6b31be\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9ee4863d7ec41439d0197f8614a4cbd : SUCCESS in 4m 22s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/7f92d42e71cf4461a4161cbc89425e73 : SUCCESS in 9m 57s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/0ce72db871d846b78c0c6539caf9a21d : SUCCESS in 8m 35s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/66957b47c086467c8eb0058c20bde67d : SUCCESS in 7m 41s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/6a219ff6265b47978de27b17a65771a7 : SUCCESS in 1h 15m 16s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/9aeb45e337dd40568b38f047d2562bfc : SUCCESS in 54m 29s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/a0f64b91e169471d986220b5eae3acb0 : SUCCESS in 1h 01m 32s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/16c022a90aa14a759732a81c14a36bbf : SUCCESS in 47m 50s","accounts_in_message":[],"_revision_number":1},{"id":"2e8f7efd98c58f20a79de96925ebfef1c2b58f3e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 11:34:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8659f7bc3fec9997b537d1b62d31ff7c2862c467","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 11:35:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9289f0adb464df4b7df2c6d9670f100\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cccb451f7c34905ad7c6590f35c71d6 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bf74084a53468e9081d7f6be161639f7b94ea0e9","revisions":{"bf74084a53468e9081d7f6be161639f7b94ea0e9":{"kind":"REWORK","_number":1,"created":"2023-08-23 08:07:45.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/82/892482/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/82/892482/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/892482/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/892482/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/82/892482/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/82/892482/1"}}},"commit":{"parents":[{"commit":"32d07c8a94ffb3a4e134217403df357e7924a341","subject":"Merge \"refactor: update render tabs components\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/32d07c8a94ffb3a4e134217403df357e7924a341"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-08-23 08:07:03.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-08-23 08:07:23.000000000","tz":480},"subject":"fix: fix the magic input","message":"fix: fix the magic input\n\nRemove the focus when the selected item is clicked to avoid the failure of the refresh button\n\nCloses-Bug: #2032763\nChange-Id: I50601a6b2bd9b36b0835fdb104f4d66b1707e728\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/bf74084a53468e9081d7f6be161639f7b94ea0e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/bf74084a53468e9081d7f6be161639f7b94ea0e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
