)]}'
{"id":"openstack%2Fskyline-console~893485","triplet_id":"openstack%2Fskyline-console~master~I656d853417dd5969fd5c4f226689111f9043b993","project":"openstack/skyline-console","branch":"master","topic":"component","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-09-04 15:11:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I656d853417dd5969fd5c4f226689111f9043b993","subject":"feat: update Action component","status":"MERGED","created":"2023-09-01 09:31:20.000000000","updated":"2023-09-04 15:12:46.000000000","submitted":"2023-09-04 15:11:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893485-component","meta_rev_id":"417674652b995f032dab14684d6a1fa94922c249","_number":893485,"virtual_id_number":893485,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-04 15:11:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-09-04 13:53:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-09-04 13:53:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-01 11:26:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-04 13:53:55.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"fdf60159dcff57ca01c5572a779147b1f47a7563","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-09-01 09:31:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"079577a587525aeb49193a290252e13bfdc1fdb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 11:26:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0aa69d5b01f34da0b03b93fd2c8ad149\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77e6f5ee14254e2fbc5285e3173aabef : SUCCESS in 4m 29s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/fc27680c4ec442158daae59e2bf4f741 : SUCCESS in 7m 36s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/cb91a571e84c4345b76803e918cb8790 : SUCCESS in 7m 04s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/24a3ee7add104d4b94211fb880051e54 : SUCCESS in 7m 42s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/45cc88f7e56d4cef88bcad7325eb81b7 : FAILURE in 1h 32m 52s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/07a92191caa94e2a8cb76357231bdb63 : FAILURE in 1h 14m 57s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/fb7f3423fe154cc58c4363b4eb71d0a9 : FAILURE in 1h 49m 46s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/261e999a7a2b49f48951769418cbb94a : FAILURE in 1h 33m 09s","accounts_in_message":[],"_revision_number":1},{"id":"8a8399a49fd95af6f003fe9bedf0d97c107b7d4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-09-04 01:08:25.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8e8396b7f346a3d55b1cabcf8d9b484dbe3efa96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 02:28:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/44414249bb3d4d72bbc69bbcae3976ec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db22c757dbe74ba29e3540661a0cbb31 : SUCCESS in 5m 15s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/0d237e5d83044ce0bb37157f89fae185 : SUCCESS in 6m 41s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/46cf3bdeac9f4661826ad01da885911e : SUCCESS in 4m 46s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/9c12cce86fb846f38e8fb4f0d5026739 : SUCCESS in 6m 49s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/df707bb7ba3b425697e3265298782fbb : FAILURE in 1h 18m 16s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/9d1c7437e0eb499483d840f36cb3763e : SUCCESS in 1h 11m 33s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/249228da0ed94868b60259e83c55bda8 : SUCCESS in 58m 29s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/1c7cd2e7674f49e884e17709c8f398f1 : SUCCESS in 40m 30s","accounts_in_message":[],"_revision_number":2},{"id":"61fdf32895d8c15fac6eb9777c30a93610d0a48c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-09-04 02:54:10.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"eb0edf6db097752fa0fa696c9f3cd69f17390e32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-09-04 03:10:59.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6ff732c9aade33687e8f4b93fcf674ba40355718","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 05:00:01.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0432b7a63d904fcfb41779e8f72c156d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90d22e82dc86451c81ad6aeb3d45970b : SUCCESS in 6m 15s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/cff1c0ba69994f72b3a51bd58c26bef5 : SUCCESS in 11m 16s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/bd96180e2e2b4678afc4d597d29b8d64 : SUCCESS in 6m 44s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/b29578b8b17f415f99f32f61c5387706 : SUCCESS in 8m 03s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/9567f6a6f1a34055925b722c5b912d1e : FAILURE in 1h 41m 20s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/8cafe28d3a4a4e92be363d3b86f88516 : SUCCESS in 1h 04m 38s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/c943e3f8c4504f4da75bc975cb16fa1d : SUCCESS in 58m 58s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/f9d8b2d6796e4af0b2e8ba3403c1f6f3 : SUCCESS in 43m 45s","accounts_in_message":[],"_revision_number":4},{"id":"e9df6dfd797854b16f64e6f68023bd64a10fec23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-09-04 10:09:15.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"42ee79deb41df6ca87d229812d09ad3e054a69aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 11:22:37.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51b6c93242844fb98210a851fc5a02ba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bec724573f37480b99bfc052b1f869ea : SUCCESS in 5m 19s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/f57b6332a7a04f6ebcb4e80cc21c8098 : SUCCESS in 9m 24s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/f943c90026354fdb93175c71e7908c7a : SUCCESS in 7m 07s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/14ac89d9e7bd47f781ef7b455cf7b181 : SUCCESS in 7m 30s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/2adf83eee41f498e9ee1e1a49058c784 : SUCCESS in 1h 11m 10s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/2bc13781f9414e23bd72a3211b60376b : SUCCESS in 49m 11s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/db398a0e3da14d688d63f3d425a21f72 : SUCCESS in 58m 52s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/7b43f37d4d164e0293961b287f7f2166 : SUCCESS in 56m 21s","accounts_in_message":[],"_revision_number":5},{"id":"8accc4c9722aeb096a51200a20c561b0af67ce83","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-09-04 13:53:55.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"5a3d38177c69711366dedf4bed33b111cd46e2ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 13:54:33.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"22093e4027854e17c3dc263f455b80fbe015f25c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 15:11:48.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a47ead3ac1945f0b2fa28bac12b25a8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73b9d30edf8840a99cda8e22d6177182 : SUCCESS in 4m 32s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/244e836b13a84e4db8409a488d773b8f : SUCCESS in 9m 58s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/0f7d798284724e10a04e23280926b742 : SUCCESS in 7m 59s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/983afa5dd5d148f08d3aaf1b8ccc02da : SUCCESS in 5m 45s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/26930debb09a48de9a2a4bc62b284397 : SUCCESS in 1h 14m 39s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/23e1845a745c4ce09e177d964d01e390 : SUCCESS in 1h 06m 48s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/11a99705ed3a46a8a7ffeabf13a0c204 : SUCCESS in 56m 23s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/74eb1d52f86443499dd6a2e56c82cd69 : SUCCESS in 42m 11s","accounts_in_message":[],"_revision_number":5},{"id":"984dc66d78f6935e7d8e8506112ddfafd4d76041","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 15:11:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"417674652b995f032dab14684d6a1fa94922c249","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 15:12:46.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1739854415149ddbb0ca543fa4dc896\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be438803a7b34ad19e9471ff185f96f0 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"94b2e8d58ff2ffe5d322eddd77d4d4004d0fc407","revisions":{"d2d29cfb52443ce6dc1003198ce1d7654a12c717":{"kind":"REWORK","_number":1,"created":"2023-09-01 09:31:20.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/85/893485/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/85/893485/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/893485/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/893485/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/893485/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/85/893485/1"}}},"commit":{"parents":[{"commit":"bf74084a53468e9081d7f6be161639f7b94ea0e9","subject":"fix: fix the magic input","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/bf74084a53468e9081d7f6be161639f7b94ea0e9"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-09-01 09:30:51.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-09-01 09:30:51.000000000","tz":480},"subject":"feat: update Action component","message":"feat: update Action component\n\nupdate Action component to check allowed by other func first\n\nChange-Id: I656d853417dd5969fd5c4f226689111f9043b993\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d2d29cfb52443ce6dc1003198ce1d7654a12c717"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d2d29cfb52443ce6dc1003198ce1d7654a12c717"}]},"branch":"refs/heads/master"},"e909f0f5d7618944ce0413f0934782d3ca70b448":{"kind":"REWORK","_number":2,"created":"2023-09-04 01:08:25.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/85/893485/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/85/893485/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/893485/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/893485/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/893485/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/85/893485/2"}}},"commit":{"parents":[{"commit":"bf74084a53468e9081d7f6be161639f7b94ea0e9","subject":"fix: fix the magic input","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/bf74084a53468e9081d7f6be161639f7b94ea0e9"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-09-01 09:30:51.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-09-04 01:06:59.000000000","tz":480},"subject":"feat: update Action component","message":"feat: update Action component\n\nupdate Action component to check allowed by other func first\n\nChange-Id: I656d853417dd5969fd5c4f226689111f9043b993\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e909f0f5d7618944ce0413f0934782d3ca70b448"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e909f0f5d7618944ce0413f0934782d3ca70b448"}]},"branch":"refs/heads/master"},"055b3a19aee998bd1a971c6e199a97b781227b76":{"kind":"REWORK","_number":3,"created":"2023-09-04 02:54:10.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/85/893485/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/85/893485/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/893485/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/893485/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/893485/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/85/893485/3"}}},"commit":{"parents":[{"commit":"bf74084a53468e9081d7f6be161639f7b94ea0e9","subject":"fix: fix the magic input","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/bf74084a53468e9081d7f6be161639f7b94ea0e9"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-09-01 09:30:51.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-09-04 02:12:12.000000000","tz":480},"subject":"feat: update Action component","message":"feat: update Action component\n\nupdate Action component to check allowed by other func first\n\nChange-Id: I656d853417dd5969fd5c4f226689111f9043b993\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/055b3a19aee998bd1a971c6e199a97b781227b76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/055b3a19aee998bd1a971c6e199a97b781227b76"}]},"branch":"refs/heads/master"},"138301836a2afa3ee09ab0ca6d4b23bd9657803e":{"kind":"REWORK","_number":4,"created":"2023-09-04 03:10:59.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/85/893485/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/85/893485/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/893485/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/893485/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/893485/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/85/893485/4"}}},"commit":{"parents":[{"commit":"bf74084a53468e9081d7f6be161639f7b94ea0e9","subject":"fix: fix the magic input","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/bf74084a53468e9081d7f6be161639f7b94ea0e9"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-09-01 09:30:51.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-09-04 03:10:30.000000000","tz":480},"subject":"feat: update Action component","message":"feat: update Action component\n\nupdate Action component to check allowed by other func first\n\nChange-Id: I656d853417dd5969fd5c4f226689111f9043b993\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/138301836a2afa3ee09ab0ca6d4b23bd9657803e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/138301836a2afa3ee09ab0ca6d4b23bd9657803e"}]},"branch":"refs/heads/master"},"94b2e8d58ff2ffe5d322eddd77d4d4004d0fc407":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-09-04 10:09:15.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/85/893485/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/85/893485/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/893485/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/893485/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/893485/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/85/893485/5"}}},"commit":{"parents":[{"commit":"e1893756d68c606d1ba014593ee90335c1df2509","subject":"fix: remove reboot e2e case","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e1893756d68c606d1ba014593ee90335c1df2509"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-09-01 09:30:51.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-09-04 10:09:15.000000000","tz":0},"subject":"feat: update Action component","message":"feat: update Action component\n\nupdate Action component to check allowed by other func first\n\nChange-Id: I656d853417dd5969fd5c4f226689111f9043b993\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/94b2e8d58ff2ffe5d322eddd77d4d4004d0fc407"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/94b2e8d58ff2ffe5d322eddd77d4d4004d0fc407"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
