)]}'
{"id":"openstack%2Fskyline-console~893594","triplet_id":"openstack%2Fskyline-console~master~I046cb196688c245cd7a3a4071e9bf2905c923ad3","project":"openstack/skyline-console","branch":"master","topic":"e2e","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-09-04 09:25:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I046cb196688c245cd7a3a4071e9bf2905c923ad3","subject":"fix: remove reboot e2e case","status":"MERGED","created":"2023-09-04 04:49:51.000000000","updated":"2023-09-04 09:26:49.000000000","submitted":"2023-09-04 09:25:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893594-e2e","meta_rev_id":"b4240181cbb62ecde79bb47e543bba52dd35783a","_number":893594,"virtual_id_number":893594,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-04 09:25:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-09-04 08:19:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-09-04 08:19:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-04 06:10:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-04 08:19:45.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"42229144b0ac990a9036deb6b0152093ab37a54c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-09-04 04:49:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08e126ef5cbee6abb7dddf690e0a523940a7cf82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 06:10:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/038c2e3842bf4676a17cd5ad9a7015f6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dbcc55d53814b5e822a8829e3e089d4 : SUCCESS in 4m 39s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/ce1dde64996c4939b0ebdede350fd152 : SUCCESS in 10m 59s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/9c84e4e660a048b0b021bb9c936c1fbc : SUCCESS in 7m 43s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/8d54bfcfa5b4429db1eb97f407a39441 : SUCCESS in 8m 14s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/cabc9a4d4a2a4ae4966b71f1f4618933 : FAILURE in 1h 18m 39s","accounts_in_message":[],"_revision_number":1},{"id":"fe996a1f988a2155fd15cfdfda24d00b0c93aacc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-09-04 06:24:57.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e943d6e9228eb6e1cea1e19a055f5d735c391325","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 07:39:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/088472f7753e4925b9f342d6a8e1d8dc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/595c4389978e421c972851a4bdd229fa : SUCCESS in 5m 01s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/dacb6e57a5a74f6988821d2b77b6467c : SUCCESS in 11m 05s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/ab063777ac63450e8585c52ef9e691d7 : SUCCESS in 7m 48s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/59ae426d4ffa42fe95dedafc1cf814c7 : SUCCESS in 5m 34s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/c57221c6490f45bf9ce7911465ebc9a1 : SUCCESS in 1h 13m 00s","accounts_in_message":[],"_revision_number":2},{"id":"ecc1d8ae5a9a4887b2f8f7cefc1780be8f55d18d","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-09-04 08:19:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3f38ef40db19faff53827dd036b6930a32da2acb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 08:20:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"51e1d9ed4a72fd05bfe4604e7229c72bcf91ba0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 09:25:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35ddc2bd7d5d4b728284b2a18f19610b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b5040c610f3488da8a579dc76369e90 : SUCCESS in 4m 20s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/bf1eb5b005e94318bcbc81c20ee0027c : SUCCESS in 8m 52s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/7e4340afcf4c4184bbdb0d2f372aefdb : SUCCESS in 5m 53s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/70ac4e88aeee48b8a5cb2a4e271a810e : SUCCESS in 6m 19s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/acf59cf242af4d67b3a3d650a86c1786 : SUCCESS in 1h 04m 33s","accounts_in_message":[],"_revision_number":2},{"id":"1b8f1b53c122da4d9207fbc926618c99efe0ed53","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 09:25:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b4240181cbb62ecde79bb47e543bba52dd35783a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 09:26:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b67fc485fc0e4cdaad358a4f1a216a35\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a647d864cae94211a9f0133075e77b19 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e1893756d68c606d1ba014593ee90335c1df2509","revisions":{"38ae8a90d018271f62b1c51c3c357652add8d003":{"kind":"REWORK","_number":1,"created":"2023-09-04 04:49:51.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/94/893594/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/94/893594/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/94/893594/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/94/893594/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/94/893594/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/94/893594/1"}}},"commit":{"parents":[{"commit":"bf74084a53468e9081d7f6be161639f7b94ea0e9","subject":"fix: fix the magic input","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/bf74084a53468e9081d7f6be161639f7b94ea0e9"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-09-04 04:49:18.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-09-04 04:49:18.000000000","tz":480},"subject":"fix: remove reboot e2e case","message":"fix: remove reboot e2e case\n\nremove reboot e2e case, for the devstack reboot instance always get error instance\n\nChange-Id: I046cb196688c245cd7a3a4071e9bf2905c923ad3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/38ae8a90d018271f62b1c51c3c357652add8d003"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/38ae8a90d018271f62b1c51c3c357652add8d003"}]},"branch":"refs/heads/master"},"e1893756d68c606d1ba014593ee90335c1df2509":{"kind":"REWORK","_number":2,"created":"2023-09-04 06:24:57.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/94/893594/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/94/893594/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/94/893594/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/94/893594/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/94/893594/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/94/893594/2"}}},"commit":{"parents":[{"commit":"bf74084a53468e9081d7f6be161639f7b94ea0e9","subject":"fix: fix the magic input","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/bf74084a53468e9081d7f6be161639f7b94ea0e9"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-09-04 04:49:18.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-09-04 06:22:57.000000000","tz":480},"subject":"fix: remove reboot e2e case","message":"fix: remove reboot e2e case\n\nremove reboot and soft reboot e2e cases, for the devstack reboot instance always get error instance\n\nChange-Id: I046cb196688c245cd7a3a4071e9bf2905c923ad3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e1893756d68c606d1ba014593ee90335c1df2509"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e1893756d68c606d1ba014593ee90335c1df2509"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
