)]}'
{"id":"openstack%2Fskyline-console~893844","triplet_id":"openstack%2Fskyline-console~master~I75c71666a32e4aee28a4d12c94e5e3eec629a98d","project":"openstack/skyline-console","branch":"master","topic":"component","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-09-06 12:40:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I75c71666a32e4aee28a4d12c94e5e3eec629a98d","subject":"feat: update Confirm component","status":"MERGED","created":"2023-09-06 09:01:19.000000000","updated":"2023-09-06 12:41:26.000000000","submitted":"2023-09-06 12:40:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893844-component","meta_rev_id":"fe04025952ac462ccc0c63341235ad1ff713c030","_number":893844,"virtual_id_number":893844,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-06 12:40:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-09-06 10:54:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-09-06 10:54:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-06 10:36:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-06 10:54:18.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"fdb4b1fc93ecdac082b395f44713c3b03243b5a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-09-06 09:01:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"313ecca47d689f3f40604918c450dcf3f3b3c4c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 10:36:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40064d7b338642c8a780a3f9b573e7a2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1b2c56ac3f542da94cc41eff3894e0e : SUCCESS in 4m 09s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/ea90f9fae7414b8ea1bdb60c0ece85e6 : SUCCESS in 11m 00s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/075cdda8c1314b899f5ceedad1a3c21b : SUCCESS in 5m 57s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/b990c481183948d6ad2c892545c74ab3 : SUCCESS in 7m 42s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/29e2c45e75a04740b398bdb00a554cd7 : SUCCESS in 1h 29m 04s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/f43fb54bd6e84eb38ce1f15771fa20c9 : SUCCESS in 1h 11m 08s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/37d6cde48c5c49a89520d21432346837 : SUCCESS in 46m 34s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/f865f84253734ca3a00658506c56d1f4 : SUCCESS in 1h 01m 49s","accounts_in_message":[],"_revision_number":1},{"id":"820cb9a4acddf672e8f2fdeb095b75aef44e6531","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-09-06 10:54:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1cd95c4505ae9dbc42f7bc6b5894773ccf7571c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 10:54:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bf3a73218eadc11d42b9e16ebd8ab1627cd0c7fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 12:40:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5b31800f7d54bb2b0a2df448d448d62\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b6386010d9d47d6a5e3e9eed685b9bd : SUCCESS in 4m 39s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/e7e4629659024c7aa0c82f7a973e2ca4 : SUCCESS in 8m 44s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/71a2a540ae6f469eadb4bde2c985dfdf : SUCCESS in 7m 19s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/72ed1edbba1f487aabcce49a04cb9d74 : SUCCESS in 6m 08s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/6fa2c2bfcbc54b1bb7137d3d5f743ab6 : SUCCESS in 1h 38m 38s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/9374d5cd6029452cac938c1f5018dfee : SUCCESS in 1h 12m 49s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/fbafcfe9d92240e1ac74a39284e70859 : SUCCESS in 59m 10s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/acf0173114a74d1fb0d9c28a0b937076 : SUCCESS in 46m 22s","accounts_in_message":[],"_revision_number":1},{"id":"ed0711796e598fa4508e6fa9e2b2d82e7a45aaff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 12:40:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fe04025952ac462ccc0c63341235ad1ff713c030","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 12:41:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51d77b41e8e346d2b97f5ecb28a31cc3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec6838af164a43bba459b0dfef4fdd89 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"19e1de7760ea7bb25c839dec1dbfba6f8e0accd2","revisions":{"19e1de7760ea7bb25c839dec1dbfba6f8e0accd2":{"kind":"REWORK","_number":1,"created":"2023-09-06 09:01:19.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/44/893844/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/44/893844/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/44/893844/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/44/893844/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/44/893844/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/44/893844/1"}}},"commit":{"parents":[{"commit":"de2793ddfbffbbe264d15f8e338ac95bc42ee270","subject":"Merge \"fix: fix table height\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/de2793ddfbffbbe264d15f8e338ac95bc42ee270"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-09-06 08:59:17.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-09-06 09:01:02.000000000","tz":480},"subject":"feat: update Confirm component","message":"feat: update Confirm component\n\nupdate Confirm component to support ok/cancel button props\n\nChange-Id: I75c71666a32e4aee28a4d12c94e5e3eec629a98d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/19e1de7760ea7bb25c839dec1dbfba6f8e0accd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/19e1de7760ea7bb25c839dec1dbfba6f8e0accd2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
