)]}'
{"id":"openstack%2Fskyline-console~893850","triplet_id":"openstack%2Fskyline-console~master~I07ba1e7b3ef7c7ef998aacf54bd5329d01737d0e","project":"openstack/skyline-console","branch":"master","topic":"component","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-09-06 12:40:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I07ba1e7b3ef7c7ef998aacf54bd5329d01737d0e","subject":"feat: update ActionButton component","status":"MERGED","created":"2023-09-06 09:28:00.000000000","updated":"2023-09-06 12:42:15.000000000","submitted":"2023-09-06 12:40:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893850-component","meta_rev_id":"d78829af7008ce6bdc21f3b0c54b4b4d42b0e771","_number":893850,"virtual_id_number":893850,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-06 12:40:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-09-06 10:54:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-09-06 10:54:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-06 10:48:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-06 10:54:30.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"28f54c21478a5da2ac60545fc9d67335ef068f13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-09-06 09:28:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fee950185bc0605fbb10c0c22cb9862c7c72b69a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 10:48:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5b52c1e193e47c9b1ef7aeef75cfffc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e32580e134534f049db183ba13321175 : SUCCESS in 4m 25s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/0870340b2a404b5d9715b5a19126fc42 : SUCCESS in 9m 10s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/8595017a0b7449b4a19335e8a05167ac : SUCCESS in 6m 20s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/026fd43498c24c729210e4e161275deb : SUCCESS in 7m 18s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/b4f57ca2b3674af99987608c4368cbfb : SUCCESS in 1h 11m 40s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/f4eb6917f25043de88e88e742382b2a6 : SUCCESS in 1h 11m 04s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/cae4bf265eb1462abc623f17ad909ef1 : SUCCESS in 41m 06s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/506fcdf644e64b5e83742bccb88f75ac : SUCCESS in 58m 01s","accounts_in_message":[],"_revision_number":1},{"id":"107432fea153a29a91dd1093ed5d2957850e8cec","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-09-06 10:54:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7dbfbb96b9fccd00aceb52fc7ab4865c10571621","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 10:55:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0c6901ba5ba352655aa37c64b495d93ba276667c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 12:40:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca17a82435a84e3190349ef993e8f8c2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccb384d8a7ed4db0b73fdae8a5eafde4 : SUCCESS in 5m 03s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/95f4f62af92f4c11807f8dcbc1e402a8 : SUCCESS in 10m 39s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/74d0eac3191f4f4596bba552a183b43a : SUCCESS in 7m 35s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/e278439b52f8404baf4dc464f4eab9e2 : SUCCESS in 7m 44s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/8dd5c1dfb072439eb0d6a81ba4f20d26 : SUCCESS in 1h 09m 28s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/f8cfc4f4967441769b66d697814d701f : SUCCESS in 1h 03m 43s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/47cde42a574648428fb4e021ad3c8a22 : SUCCESS in 1h 02m 19s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/e0b27f3464e1452eb5a4b87ca4d08fe8 : SUCCESS in 56m 49s","accounts_in_message":[],"_revision_number":1},{"id":"1d531c36c80860a4a41507d8480d4f0c41da8f34","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 12:40:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d78829af7008ce6bdc21f3b0c54b4b4d42b0e771","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 12:42:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccb005646ba64cf38adce1cd0c2c7b99\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acef38f7624f4c64a70e99bb5eabfce0 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"74ffcc6215475c4e612463efd0968d591d48159a","revisions":{"74ffcc6215475c4e612463efd0968d591d48159a":{"kind":"REWORK","_number":1,"created":"2023-09-06 09:28:00.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/50/893850/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/50/893850/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/50/893850/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/50/893850/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/50/893850/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/50/893850/1"}}},"commit":{"parents":[{"commit":"19e1de7760ea7bb25c839dec1dbfba6f8e0accd2","subject":"feat: update Confirm component","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/19e1de7760ea7bb25c839dec1dbfba6f8e0accd2"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-09-06 09:27:04.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-09-06 09:27:04.000000000","tz":480},"subject":"feat: update ActionButton component","message":"feat: update ActionButton component\n\n1. refactor ActionButton\n2. support ok/cancel button props for confirm/modal action\n\nChange-Id: I07ba1e7b3ef7c7ef998aacf54bd5329d01737d0e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/74ffcc6215475c4e612463efd0968d591d48159a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/74ffcc6215475c4e612463efd0968d591d48159a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
