)]}'
{"id":"openstack%2Fskyline-console~893998","triplet_id":"openstack%2Fskyline-console~master~Ib22120105deb30f108218d040fc1fc33f88b39bc","project":"openstack/skyline-console","branch":"master","topic":"component","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-09-07 09:06:20.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}},"hashtags":[],"change_id":"Ib22120105deb30f108218d040fc1fc33f88b39bc","subject":"feat: add action props to ModalAction component","status":"MERGED","created":"2023-09-07 02:26:50.000000000","updated":"2023-09-07 11:41:47.000000000","submitted":"2023-09-07 11:40:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893998-component","meta_rev_id":"d973aac059a1492dcb20e9486498d5ef27fc2481","_number":893998,"virtual_id_number":893998,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-07 11:40:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-09-07 07:31:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-09-07 07:31:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-07 04:19:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-07 07:31:59.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"452584e90e00d9e8bc7ac3ddcc72696867acd763","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-09-07 02:26:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8222f707da70bb00cae01924f7894fb49982df06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 04:19:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3abf27e2793d4647a9216ac9682b7466\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d66dccaff91d43208b95e3e9b1bc1634 : SUCCESS in 5m 12s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/94163dccb6c648f9842142d7035bc6ce : SUCCESS in 12m 00s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/1c162b2560bf40399a3d120ce1905c66 : SUCCESS in 8m 46s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/eebe37219b1546758ca33056f043fb34 : SUCCESS in 7m 51s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/c9a70392de71474ea208ca6ad904dd3e : SUCCESS in 1h 21m 40s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/850944b41ef44e29a1411096ac227e8a : SUCCESS in 1h 19m 41s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/0135730addfa4590b3e09f2e62083373 : SUCCESS in 59m 32s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/37fd185730c54fa093dabab1a0f05c8f : FAILURE in 1h 26m 30s","accounts_in_message":[],"_revision_number":1},{"id":"5c18c4c964312c69ef6fe7a6cef8fb6ba0ede6a2","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-09-07 06:11:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"23b9eb48dc4641ccf778c380d59e152eb5e6820e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 07:20:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86056f124acc48a99f980dcb141bd66d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a725e4ba7e9c430c82d363f2873f9f42 : SUCCESS in 4m 46s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/3421c06c383b4058abe2a215177ee7c4 : SUCCESS in 10m 10s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/b342c3ce852d44b3b2c9fec0ed29afdf : SUCCESS in 6m 53s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/b85af6b4b01b4342a610da1d5e755833 : SUCCESS in 6m 09s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/95f9aa4b1eac4045ad62376fed3a82ae : SUCCESS in 1h 03m 28s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/d927d58fe8864cdebef9fa78bfc4c4c0 : SUCCESS in 1h 03m 52s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/cb2b90c3a2ce4da48a971e74fe9db4b3 : SUCCESS in 59m 28s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/0cb23fb132da4c0f99cafc3dd101d87c : SUCCESS in 56m 42s","accounts_in_message":[],"_revision_number":1},{"id":"aa02e781463fe00b4206683801d244a69f6b5a8d","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-09-07 07:31:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9da1763cf9dc9c475d190e4294f20beb326b1676","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 07:32:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e9cecea8c030fb79fe4c732d2ee7a2e6c4c233b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 09:04:22.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6bfcc65449f44ea487445846630e1be1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0f1ab36abf14da79025882eec7929bc : SUCCESS in 4m 24s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/4e902241c0944df0810047ed60aab891 : SUCCESS in 11m 09s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/11e481e663cd4899803fd66c88263f8e : SUCCESS in 5m 00s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/6358a355ac4348ccb0264bf84097c842 : SUCCESS in 4m 54s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/f07ced39ee8e48f5a0245405a9620bde : FAILURE in 1h 27m 07s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/4c2832249548435cbc48e9e22e4052be : SUCCESS in 1h 11m 44s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/bbcd578051c44284bc411b1be5b276ed : SUCCESS in 1h 04m 29s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/0391550262574d2c8b9f5a67cc137ed1 : SUCCESS in 59m 22s","accounts_in_message":[],"_revision_number":1},{"id":"98d3b9efdde57b97c63ae00ba737afd376042af3","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-09-07 09:06:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0ffdd94895f84663268c0e3a1300a93d42331be1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 10:20:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ccc05061d904f7394f664d4fd1e0b79\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e4dd89df75b425f9e416aedce37660d : SUCCESS in 4m 01s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/cbec94afde8e4548a89cd291e8a25a4a : SUCCESS in 8m 05s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/de107c5e345841688117aee17fd2e32b : SUCCESS in 5m 56s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/b286758f80554e6b92191b4900203658 : SUCCESS in 7m 13s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/31f0ca14cc2044e192295482a7874833 : SUCCESS in 1h 12m 24s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/ea8c46716a2f4a6f8958e65ac88e2f64 : SUCCESS in 52m 41s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/9c4d78bd4173455e968c3639e133b990 : SUCCESS in 47m 56s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/637f0a0c61634918b20d7321f6153c67 : SUCCESS in 52m 10s","accounts_in_message":[],"_revision_number":1},{"id":"79d381a1090c7bdd2ac89bd728d58d08819de7d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 10:20:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b8b63d386789dd057161c277ce4133cadf9a974b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 11:40:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84c1618772a94052af2d3099c281bc18\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9fd32912eb24d60aa9c7d904807d759 : SUCCESS in 5m 16s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/b46473c4b2564ded9e12b4efe63d4190 : SUCCESS in 9m 03s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/cb612fe55ffb45f4ae3d7a969f340998 : SUCCESS in 7m 00s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/1cee382e1b824643b04fc59bebcef692 : SUCCESS in 6m 49s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/e14df43986a04991b65e6efad2695ab7 : SUCCESS in 1h 18m 50s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/4962eabdba2543e1bef7d14801545500 : SUCCESS in 1h 06m 42s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/593ab825df0f4397af54363dac12cb06 : SUCCESS in 1h 05m 24s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/fd94185ed6c0484a8bd87534f7f0a4d3 : SUCCESS in 55m 58s","accounts_in_message":[],"_revision_number":1},{"id":"85c0c69e2740f7bc2d1e325d491bd9250ab0777a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 11:40:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d973aac059a1492dcb20e9486498d5ef27fc2481","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 11:41:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fcfc79c491d4803b18a271e1dbeb3fb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1808ae7b0dea43faa20b25e24cf0ff65 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d5b58ac16dfe20b6173893c92937aa102abdee40","revisions":{"d5b58ac16dfe20b6173893c92937aa102abdee40":{"kind":"REWORK","_number":1,"created":"2023-09-07 02:26:50.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/98/893998/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/98/893998/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/98/893998/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/98/893998/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/98/893998/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/98/893998/1"}}},"commit":{"parents":[{"commit":"74ffcc6215475c4e612463efd0968d591d48159a","subject":"feat: update ActionButton component","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/74ffcc6215475c4e612463efd0968d591d48159a"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-09-07 02:25:18.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-09-07 02:25:18.000000000","tz":480},"subject":"feat: add action props to ModalAction component","message":"feat: add action props to ModalAction component\n\nadd action props to ModalAction component, then can use the action static props for the instance function\n\nChange-Id: Ib22120105deb30f108218d040fc1fc33f88b39bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d5b58ac16dfe20b6173893c92937aa102abdee40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d5b58ac16dfe20b6173893c92937aa102abdee40"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
