)]}'
{"id":"openstack%2Fskyline-console~894306","triplet_id":"openstack%2Fskyline-console~master~I4e11cd7a3461c20678d142bfdd4e9a27c95af2b8","project":"openstack/skyline-console","branch":"master","topic":"I74913bbf1c48823ba4a9b68539a2c26582e62939","attention_set":{},"removed_from_attention_set":{"36242":{"account":{"_account_id":36242,"name":"omar asim mirza","email":"omarasimmirza@gmail.com","username":"omar.asim"},"last_update":"2023-09-13 07:52:08.000000000","reason":"Change was submitted"},"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2023-09-13 07:52:08.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-09-13 06:15:41.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2023-09-13 07:52:08.000000000","reason":"Change was submitted"},"34671":{"account":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"last_update":"2023-09-13 07:52:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4e11cd7a3461c20678d142bfdd4e9a27c95af2b8","subject":"admin-menu.jsx updated.","status":"MERGED","created":"2023-09-08 14:27:05.000000000","updated":"2023-09-13 07:53:53.000000000","submitted":"2023-09-13 07:52:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"894306-I74913bbf1c48823ba4a9b68539a2c26582e62939","meta_rev_id":"c1bf9a82e3e7a9b7f7f6cbca39611f86189633d2","_number":894306,"virtual_id_number":894306,"owner":{"_account_id":36242,"name":"omar asim mirza","email":"omarasimmirza@gmail.com","username":"omar.asim"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36242,"name":"omar asim mirza","email":"omarasimmirza@gmail.com","username":"omar.asim"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-13 07:52:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"recommended":{"_account_id":36242,"name":"omar asim mirza","email":"omarasimmirza@gmail.com","username":"omar.asim"},"all":[{"value":1,"date":"2023-09-11 05:14:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36242,"name":"omar asim mirza","email":"omarasimmirza@gmail.com","username":"omar.asim"},{"value":2,"date":"2023-09-13 06:15:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":0,"_account_id":36242,"name":"omar asim mirza","email":"omarasimmirza@gmail.com","username":"omar.asim"},{"value":1,"date":"2023-09-13 06:15:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"}],"reviewers":{"CC":[{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},{"_account_id":36242,"name":"omar asim mirza","email":"omarasimmirza@gmail.com","username":"omar.asim"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-08 14:41:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-11 05:14:41.000000000","updated_by":{"_account_id":36242,"name":"omar asim mirza","email":"omarasimmirza@gmail.com","username":"omar.asim"},"reviewer":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"state":"CC"},{"updated":"2023-09-11 05:14:41.000000000","updated_by":{"_account_id":36242,"name":"omar asim mirza","email":"omarasimmirza@gmail.com","username":"omar.asim"},"reviewer":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"state":"REVIEWER"},{"updated":"2023-09-11 05:14:41.000000000","updated_by":{"_account_id":36242,"name":"omar asim mirza","email":"omarasimmirza@gmail.com","username":"omar.asim"},"reviewer":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"state":"REVIEWER"},{"updated":"2023-09-11 05:14:41.000000000","updated_by":{"_account_id":36242,"name":"omar asim mirza","email":"omarasimmirza@gmail.com","username":"omar.asim"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2023-09-11 05:14:41.000000000","updated_by":{"_account_id":36242,"name":"omar asim mirza","email":"omarasimmirza@gmail.com","username":"omar.asim"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"b2136085ff409af3fe0e245e4d3662a03335c5c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36242,"name":"omar asim mirza","email":"omarasimmirza@gmail.com","username":"omar.asim"},"date":"2023-09-08 14:27:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"314980e59dcbfd51f02927bc065a7a1921799c06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 14:41:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/790d3c54e8154989960b7faf4a3a84c8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fa2eb2ca1484a12b397dcb0552a23bd : SUCCESS in 4m 52s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/44f1d3454e8a4ff4aa7c0a4354d41545 : SUCCESS in 9m 30s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/d50b87189ab543edb8b175e441f89d4b : SUCCESS in 8m 42s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/5f993f7c15e845e8a25ae83deea90648 : SUCCESS in 7m 27s","accounts_in_message":[],"_revision_number":1},{"id":"9f23b2932b5b3e6ff062b2fbbb677bfe7d6880bf","author":{"_account_id":36242,"name":"omar asim mirza","email":"omarasimmirza@gmail.com","username":"omar.asim"},"date":"2023-09-11 05:14:41.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c9e8bff8c5d845a27599c3f3dc0fd7e547f9acd7","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-09-13 06:15:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"805426b0ce5a2f3b62623ac40f0106bc1e5c349f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 06:16:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"03bd21f1b4c88ea40da17cdaa44a7999c4bfd4bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 07:52:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b30a2d95b924b38a9bc0f594468fd88\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70568263a646478185fda6f146b82dfa : SUCCESS in 4m 05s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/9426786f5f0a4d2f9f0dd5374c9715c9 : SUCCESS in 8m 09s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/349db4d9159f46d89cd9aa7bd19519fc : SUCCESS in 5m 02s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/2a44fba475cf43ff804c772b1128654f : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":1},{"id":"fab728a96903da43ee0b338d138c88a44dce3cc3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 07:52:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c1bf9a82e3e7a9b7f7f6cbca39611f86189633d2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 07:53:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c3d69be86e54b688c9afb7812eb3464\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2c6a9d80598453184cd439decfbf795 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8d0678db83cf7b172340f3912f8b2c8974fb7a89","revisions":{"8d0678db83cf7b172340f3912f8b2c8974fb7a89":{"kind":"REWORK","_number":1,"created":"2023-09-08 14:27:05.000000000","uploader":{"_account_id":36242,"name":"omar asim mirza","email":"omarasimmirza@gmail.com","username":"omar.asim"},"ref":"refs/changes/06/894306/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/06/894306/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/06/894306/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/06/894306/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/06/894306/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/06/894306/1"}}},"commit":{"parents":[{"commit":"0bc19114c43f16433143e34df0ecd54c004205ad","subject":"Add Masakari UI to Skyline-Console","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0bc19114c43f16433143e34df0ecd54c004205ad"}]}],"author":{"name":"omar asim mirza","email":"omarasimmirza@gmail.com","date":"2023-09-08 14:20:58.000000000","tz":0},"committer":{"name":"omar asim mirza","email":"omarasimmirza@gmail.com","date":"2023-09-08 14:22:55.000000000","tz":0},"subject":"admin-menu.jsx updated.","message":"admin-menu.jsx updated.\n\nChange-Id: I4e11cd7a3461c20678d142bfdd4e9a27c95af2b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8d0678db83cf7b172340f3912f8b2c8974fb7a89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8d0678db83cf7b172340f3912f8b2c8974fb7a89"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
